)]}'
{"id":"x%2Fpackstack~833840","triplet_id":"x%2Fpackstack~master~Id1eb2b1e4ad6f4ac2abe3dd7a7e81d555ca95d70","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-15 21:07:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id1eb2b1e4ad6f4ac2abe3dd7a7e81d555ca95d70","subject":"Cinder: Remove tenant_id templating from endpoint","status":"MERGED","created":"2022-03-15 15:23:50.000000000","updated":"2022-03-15 21:07:48.000000000","submitted":"2022-03-15 21:07:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"833840","meta_rev_id":"7cb920f807e645321fe9d74dbe1314fae547d26a","_number":833840,"virtual_id_number":833840,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-15 21:07:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2022-03-15 20:15:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2022-03-15 20:15:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-15 16:23:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-15 20:15:20.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"73797fa071b9c2efd4cc72b9ced7604812ec5cc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-15 15:23:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ff07eb3bbe37488162d83fc954a3e8e2237ef1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 16:23:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a7ad3d7bf6249678fd8cafb0e8d7b67 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88c89036a909488d86fb68aa6729cbcb : SUCCESS in 5m 43s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/2adc4e8a6197453293fcf92cfded8f91 : SUCCESS in 48m 39s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/9e421e41874443798eee3730945bfaf7 : SUCCESS in 53m 32s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/78d22b639cd4473faa3434d002e55662 : SUCCESS in 51m 56s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/2aa2efcc13d64b70b7adf5f392417916 : SUCCESS in 56m 18s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/16cc362109524abb96c1634c6ea0f216 : SUCCESS in 37m 05s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/80c7afd907e448b1ac71578cdc962320 : SUCCESS in 40m 49s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/0eeb2abbb7d34e1384b063652e5212fa : SUCCESS in 39m 52s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/cf428642c7f949a5a0a692508de3e8dc : SUCCESS in 47m 12s","accounts_in_message":[],"_revision_number":1},{"id":"01c7cbe872820a20779f7b0b74166716a8eeb355","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-03-15 20:15:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"830783b8ed3ead8ad0083f8cde0b6e1ea204d841","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 20:16:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"27ff2b6218c2daf86b989e4d5a1c89c4622ffc56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 21:07:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be298f141ea440f2ba7e6ea2a67f3095 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5dd8cd1a7944e068bd1c36777257e30 : SUCCESS in 5m 47s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/e9f240f29fc74e61bfa8219898441a53 : SUCCESS in 36m 27s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/f5ed3e3240b149e195ab61efc1c57a4f : SUCCESS in 38m 35s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/3904eea1277143db877e1ea6348c47a6 : SUCCESS in 39m 50s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/555e4cac04bb49e8be12af14492653b6 : SUCCESS in 43m 29s","accounts_in_message":[],"_revision_number":1},{"id":"7cb920f807e645321fe9d74dbe1314fae547d26a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 21:07:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f8636615169a650ae756bad99378f235e5c82ade","revisions":{"f8636615169a650ae756bad99378f235e5c82ade":{"kind":"REWORK","_number":1,"created":"2022-03-15 15:23:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/833840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/40/833840/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/40/833840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/40/833840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/40/833840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/40/833840/1"}}},"commit":{"parents":[{"commit":"174bd7ddee3ac2e666ec81693ddf313a40fe21d0","subject":"Manila: Remove tenant_id templating from v2 endpint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/174bd7ddee3ac2e666ec81693ddf313a40fe21d0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-15 15:22:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-15 15:22:15.000000000","tz":540},"subject":"Cinder: Remove tenant_id templating from endpoint","message":"Cinder: Remove tenant_id templating from endpoint\n\nCinder no longer requires tenant_id templating[1]. We should remove it\nso that system users can use API.\n\n[1] https://review.opendev.org/c/openstack/cinder/+/822040\n\nChange-Id: Id1eb2b1e4ad6f4ac2abe3dd7a7e81d555ca95d70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f8636615169a650ae756bad99378f235e5c82ade"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f8636615169a650ae756bad99378f235e5c82ade"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
