)]}'
{"id":"x%2Fpackstack~836589","triplet_id":"x%2Fpackstack~stable%2Fyoga~I9a4f94f85e95e16ff9fe60f60fbe5e599162bf7b","project":"x/packstack","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-04-11 05:54:10.000000000","reason":"yatin replied on the change"},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2022-08-16 15:00:25.000000000","reason":"\u003cGERRIT_ACCOUNT_16312\u003e replied on the change","reason_account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2022-08-16 16:03:15.000000000","reason":"Change was submitted"},"13294":{"account":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"last_update":"2022-08-16 16:03:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9a4f94f85e95e16ff9fe60f60fbe5e599162bf7b","subject":"Adapt for stable/yoga","status":"MERGED","created":"2022-04-05 10:29:41.000000000","updated":"2022-08-16 16:03:15.000000000","submitted":"2022-08-16 16:03:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"836589","meta_rev_id":"f9f4b519441c066598d547ed0f986e16b67550bd","_number":836589,"virtual_id_number":836589,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-16 16:03:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2022-08-16 12:58:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2022-08-16 15:00:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"reviewers":{"REVIEWER":[{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-05 11:32:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-05 12:01:10.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-04-05 12:01:10.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2022-04-05 12:01:10.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"}],"messages":[{"id":"d9555484cd2ca73c1d07403b71f702e45fb2e541","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-04-05 10:29:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be361883d8f7c800c385654ae9079efeb69e8246","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-04-05 10:35:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5151430a49e77a0c3aaed0f94938d74d53b42002","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 11:32:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/110dd214e9db4213a94e10afa6d5691f : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e49a76b807440599603f43878b8d160 : SUCCESS in 4m 22s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/88d68d1dc3a2494984c8b0f09559596e : SUCCESS in 39m 58s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/71287634fd674cd1aeb70098813c77bd : SUCCESS in 45m 24s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/2e684479a77843a6aa8665429b6ff1ad : SUCCESS in 49m 09s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/4117894ac8b74f2594d25c07032b0c40 : SUCCESS in 52m 39s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/31ff9f2cf6f3422eb0d0220825e13c89 : SUCCESS in 35m 52s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/57b75e9aa8f04122b8008f3eea26fc53 : SUCCESS in 38m 04s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/9302d01fe2dc4771ae10a13166f727a9 : SUCCESS in 41m 20s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ce6467ec01d7465899e1506adf4f5a25 : SUCCESS in 45m 09s","accounts_in_message":[],"_revision_number":2},{"id":"421c197a3066d89a402225d9a5f4c91c3ede27f1","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-04-11 05:54:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d92e05eb4b2b73e3b14013126b928295cf40ff50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-08-16 11:00:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e706485b77205832ca52e4f4f7c49d7d86242797","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 11:31:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9519bb434f94def91e7786dbfcd55a8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b857109007643f38f4298be4720cce2 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a335c4ee1e464d99ba38d1da3fa54e53 : SUCCESS in 4m 05s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/1e6f407173504100b9c1f410e61ba6aa : POST_FAILURE in 15m 37s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/b3ba3a01a59e494f8df31201450cf43a : POST_FAILURE in 20m 50s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/c32fe57bf60b4711992be53debaccb42 : POST_FAILURE in 18m 47s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/61889b093bc7494393f54a997b28972f : POST_FAILURE in 19m 43s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/f2b5e48a87e4435aa7917f13634aeb5f : POST_FAILURE in 18m 14s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/208fec90ff494002a2fedd12d33b7dfd : POST_FAILURE in 20m 34s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/f25c8865412a4e1985b8b39dd68223cf : POST_FAILURE in 20m 14s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/2b12eff86ff74b0ba338fbfb6093a4ca : POST_FAILURE in 25m 29s","accounts_in_message":[],"_revision_number":3},{"id":"e68805d86771e063c991b0530b60f272137f3da2","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-08-16 11:36:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"58837e4adf395cfbae0945132cd431bcf8658cae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-08-16 11:57:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f5d25c551865b8fdb376948afe2e72b6dc781b3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 12:53:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6112f66759474db39659f8d090f0b7c9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aeb7ef2a6e4405a982d9816f7cb8572 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3dbfd4385f0942dda43a06c8265ef551 : SUCCESS in 4m 39s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/67d9dbd3277a4623b14c6422c7f66e65 : SUCCESS in 44m 21s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/7cfdf516404440788e1716c666532d90 : SUCCESS in 42m 20s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/fbc8370533b9488a93cc465f22dfe4cb : SUCCESS in 47m 34s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/76b83ed02c5d442b8dc576e38ba6e864 : SUCCESS in 52m 00s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/4ad71c980ab84b349c5faaa2cc752531 : SUCCESS in 41m 49s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/bd8ae3373a0b4c00bd549a3357b79cb0 : SUCCESS in 40m 49s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/d51e4ffc3d754cde845c382811645732 : SUCCESS in 44m 15s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/22aced52bbeb49b286c0baff30781b1b : SUCCESS in 46m 23s","accounts_in_message":[],"_revision_number":4},{"id":"926f82703bc48b9e7b846a77fe4dfe93721ad40c","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-08-16 12:58:25.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c0f47fe5f5fe95218ed03cac51dd9d034a8aa3d9","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-08-16 15:00:25.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6ea1d48b2d98a314c1b18437fb2f2e497ef53872","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 15:01:14.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"467c6b8e54f5e6e8c64d1312e83f0c77180c8398","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 16:03:14.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08e2b74bf7be4576b1f39e32585af116\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1b67021dcc541b5b45dd9217733f177 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63b87c14f799414bb14d687ca872be43 : SUCCESS in 4m 32s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/0b2dc057e85d41c8ae1bc318619b2309 : SUCCESS in 43m 22s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/5ad28064470f49b582018cd958ece7d5 : SUCCESS in 43m 23s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/3d79c63de91d4d5fb26a002442707c46 : SUCCESS in 57m 59s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/986598eea54140fe99da750e0d7a79a6 : SUCCESS in 50m 11s","accounts_in_message":[],"_revision_number":4},{"id":"f9f4b519441c066598d547ed0f986e16b67550bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 16:03:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"eac89269a4ccbd496c5fc335540eb6d6f61687a1","revisions":{"48219733a982086de88de515794d536d393ceebb":{"kind":"REWORK","_number":1,"created":"2022-04-05 10:29:41.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/89/836589/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/89/836589/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/89/836589/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/89/836589/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/89/836589/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/89/836589/1"}}},"commit":{"parents":[{"commit":"f8636615169a650ae756bad99378f235e5c82ade","subject":"Cinder: Remove tenant_id templating from endpoint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f8636615169a650ae756bad99378f235e5c82ade"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-04-05 10:29:15.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-04-05 10:29:15.000000000","tz":120},"subject":"Fix .gitreview for stable/yoga","message":"Fix .gitreview for stable/yoga\n\nChange-Id: I9a4f94f85e95e16ff9fe60f60fbe5e599162bf7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/48219733a982086de88de515794d536d393ceebb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/48219733a982086de88de515794d536d393ceebb"}]},"branch":"refs/heads/stable/yoga"},"9b8c3aeabd067753d7c0c53e40677c9cdbf78776":{"kind":"REWORK","_number":2,"created":"2022-04-05 10:35:21.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/89/836589/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/89/836589/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/89/836589/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/89/836589/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/89/836589/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/89/836589/2"}}},"commit":{"parents":[{"commit":"f8636615169a650ae756bad99378f235e5c82ade","subject":"Cinder: Remove tenant_id templating from endpoint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f8636615169a650ae756bad99378f235e5c82ade"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-04-05 10:29:15.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-04-05 10:34:53.000000000","tz":120},"subject":"Fix .gitreview for stable/yoga","message":"Fix .gitreview for stable/yoga\n\nAlso using puppet openstack modules from stable/yoga and rpm repos for\nyoga instead of master.\n\nChange-Id: I9a4f94f85e95e16ff9fe60f60fbe5e599162bf7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9b8c3aeabd067753d7c0c53e40677c9cdbf78776"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9b8c3aeabd067753d7c0c53e40677c9cdbf78776"}]},"branch":"refs/heads/stable/yoga"},"416c0719d55b71be9237d9b061aaf734683bd302":{"kind":"REWORK","_number":3,"created":"2022-08-16 11:00:16.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/89/836589/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/89/836589/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/89/836589/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/89/836589/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/89/836589/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/89/836589/3"}}},"commit":{"parents":[{"commit":"f8636615169a650ae756bad99378f235e5c82ade","subject":"Cinder: Remove tenant_id templating from endpoint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f8636615169a650ae756bad99378f235e5c82ade"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-04-05 10:29:15.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-08-16 11:00:14.000000000","tz":120},"subject":"Adapt for stable/yoga","message":"Adapt for stable/yoga\n\n- Fix .gitreview\n- Using puppet openstack modules from stable/yoga\n- Using non-openstack puppet modules from pinned commit hashes in RDO\n- Using rpm repos from yoga instead of master\n\nChange-Id: I9a4f94f85e95e16ff9fe60f60fbe5e599162bf7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/416c0719d55b71be9237d9b061aaf734683bd302"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/416c0719d55b71be9237d9b061aaf734683bd302"}]},"branch":"refs/heads/stable/yoga"},"eac89269a4ccbd496c5fc335540eb6d6f61687a1":{"kind":"REWORK","_number":4,"created":"2022-08-16 11:57:24.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/89/836589/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/89/836589/4","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/89/836589/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/89/836589/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/89/836589/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/89/836589/4"}}},"commit":{"parents":[{"commit":"f8636615169a650ae756bad99378f235e5c82ade","subject":"Cinder: Remove tenant_id templating from endpoint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f8636615169a650ae756bad99378f235e5c82ade"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-04-05 10:29:15.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-08-16 11:57:10.000000000","tz":120},"subject":"Adapt for stable/yoga","message":"Adapt for stable/yoga\n\n- Fix .gitreview\n- Using puppet openstack modules from stable/yoga\n- Using non-openstack puppet modules from pinned commit hashes in RDO\n- Using rpm repos from yoga instead of master\n\nChange-Id: I9a4f94f85e95e16ff9fe60f60fbe5e599162bf7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/eac89269a4ccbd496c5fc335540eb6d6f61687a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/eac89269a4ccbd496c5fc335540eb6d6f61687a1"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[]}
