)]}'
{"id":"x%2Fpackstack~837062","triplet_id":"x%2Fpackstack~stable%2Fyoga~I5618353cb368e9d56188297dfdf87b1b127ee653","project":"x/packstack","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2022-04-08 18:32:48.000000000","reason":"Alfredo Moralejo replied on the change"}},"hashtags":[],"change_id":"I5618353cb368e9d56188297dfdf87b1b127ee653","subject":"Support CS9 when installing RDO repo","status":"MERGED","created":"2022-04-08 17:40:39.000000000","updated":"2022-04-08 19:41:27.000000000","submitted":"2022-04-08 19:41:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837062","cherry_pick_of_change":837121,"cherry_pick_of_patch_set":2,"meta_rev_id":"7e8b788109e9cece7d5c966d762762c1b7a3ba16","_number":837062,"virtual_id_number":837062,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-08 19:41:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2022-04-08 18:32:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2022-04-08 18:07:49.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2022-04-08 18:32:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-08 17:40:39.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"CC"},{"updated":"2022-04-08 18:07:49.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"},{"updated":"2022-04-08 18:29:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7b74dd13783ea0e68fedcc7409a5fca635972fc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-04-08 17:40:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"49b22f5c1bc7e0b8ccb6f344c18b721a1c000eda","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2022-04-08 18:07:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"caa44105235d178ad00081642de6a5fa929e56b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 18:29:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6503f9a240e42db82670331f87a7acb : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bfb475154cb460aa097aecb578e4f73 : SUCCESS in 3m 50s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/973967aa68714083b34019a10a708aea : SUCCESS in 37m 54s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/32865c38ec544e8ea8ef550498c1e4b9 : SUCCESS in 41m 57s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/7821db940d1847d1a7cb3fba22aef750 : SUCCESS in 36m 47s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/4fda205d9c2c4a93899cef4b44f174b3 : SUCCESS in 47m 25s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/33ba4e1383684b77bb2774c1807c60d6 : SUCCESS in 30m 20s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/25cb8c4604234e0d8d62b42bccdba768 : SUCCESS in 29m 27s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/9a15049cb417425ea369f8d0132862b7 : SUCCESS in 33m 53s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/4f3865ea9d4a493c86de5f190573da9f : SUCCESS in 37m 59s","accounts_in_message":[],"_revision_number":1},{"id":"f7ea9beb3bd0812d50e11b8ace3d6889f2c1c803","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-04-08 18:32:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c10189990ef4129f604645b909fe2fd07d91b13f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 18:33:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"091df7fc64c75a1f3b8770b45d1500b3cf7bf624","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 19:41:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56210dad2ad845948d989389e3723bdd : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a794c0fe7e340a4bfb4dd146166ecaf : SUCCESS in 4m 15s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/f632d74b8fe74a73978bdd381652f8f7 : SUCCESS in 1h 02m 40s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/6270fd4eca7841d3bc318d4aaca67b86 : SUCCESS in 54m 54s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/562d88dc76684635968044b3ab012907 : SUCCESS in 1h 03m 36s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9572493fedea4e63a717daadd9e3f097 : SUCCESS in 38m 36s","accounts_in_message":[],"_revision_number":1},{"id":"7e8b788109e9cece7d5c966d762762c1b7a3ba16","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 19:41:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5e7188e5e13bc41296e90062407725054eb4e488","revisions":{"5e7188e5e13bc41296e90062407725054eb4e488":{"kind":"REWORK","_number":1,"created":"2022-04-08 17:40:39.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/62/837062/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/62/837062/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/62/837062/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/62/837062/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/62/837062/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/62/837062/1"}}},"commit":{"parents":[{"commit":"f8636615169a650ae756bad99378f235e5c82ade","subject":"Cinder: Remove tenant_id templating from endpoint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f8636615169a650ae756bad99378f235e5c82ade"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2022-04-08 12:43:37.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-04-08 17:40:39.000000000","tz":0},"subject":"Support CS9 when installing RDO repo","message":"Support CS9 when installing RDO repo\n\nChange-Id: I5618353cb368e9d56188297dfdf87b1b127ee653\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5e7188e5e13bc41296e90062407725054eb4e488"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5e7188e5e13bc41296e90062407725054eb4e488"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
