)]}'
{"id":"x%2Fpackstack~837121","triplet_id":"x%2Fpackstack~master~I5618353cb368e9d56188297dfdf87b1b127ee653","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2022-04-08 19:17:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5618353cb368e9d56188297dfdf87b1b127ee653","subject":"Support CS9 when installing RDO repo","status":"MERGED","created":"2022-04-08 12:44:10.000000000","updated":"2022-04-08 19:17:21.000000000","submitted":"2022-04-08 19:17:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837121","meta_rev_id":"565e0f1e4d3abe7576438e3701eba64a49dff341","_number":837121,"virtual_id_number":837121,"owner":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-08 19:17:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2022-04-08 18:05:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2022-04-08 18:05:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-08 12:58:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-08 17:28:51.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"CC"},{"updated":"2022-04-08 18:05:43.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"9ebf35bd38b7a9e562e92c4bf21e23321392617b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2022-04-08 12:44:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8791fb65f2746ab04cadc62d9637c1623950884e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 12:58:32.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c59bccb64e944029479d2663ad5b492 : FAILURE in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0ec040949634c5881d60b957235b36b : FAILURE in 3m 45s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/2f9c048e3e5549a18c678ecca357015f : FAILURE in 11m 49s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/14fede7484c74577b47fc504b43c5ae7 : FAILURE in 11m 56s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/768e9f425259442e93798d4a07363df6 : FAILURE in 11m 27s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c9f01fb8d6b14c7a98332fbd928bee9d : FAILURE in 12m 18s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/7bbf767867844bbea65692e909c30e97 : FAILURE in 9m 18s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/0262b58e2fb8413fb5584480dbb3fb4c : FAILURE in 9m 26s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/3b26023bfc6241f4a25ae386be4dc5d5 : FAILURE in 10m 40s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f71cbae5f5ac4c0ea81ccc03b4cf8261 : FAILURE in 11m 39s","accounts_in_message":[],"_revision_number":1},{"id":"0bde912e8ca70d62c0a95c234ef501c0430a5eef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2022-04-08 13:19:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7ea32e4b472a26d4e1144622e274e843d8f992c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 14:25:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de79ff82bee34d43a36b9e246c6c5d04 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20440982e70640fabd12c92e38807b1d : SUCCESS in 3m 26s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/b56ca6d0439a428fb61a1ac8be4fc804 : SUCCESS in 44m 04s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/15cf8938a442458b9af9a9a7b22f7270 : SUCCESS in 1h 00m 14s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/5e42f454a94740a481ac27332e7d2ca2 : SUCCESS in 49m 50s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d2654e52ef4b455eaf96c8cc715e6062 : POST_FAILURE in 6m 28s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/18360f423bf64df8ab88f68c41067d84 : SUCCESS in 33m 45s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/80e57f8482294bccb0dd4a08510edcd9 : SUCCESS in 40m 08s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/fa1e47f14fb54c89934d154274238aca : SUCCESS in 35m 10s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/93d7638129a94f5f9f8842d16cbc38b8 : SUCCESS in 45m 14s","accounts_in_message":[],"_revision_number":2},{"id":"ec42769d2422c1d86e68879bf67b9ba42a526b32","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2022-04-08 15:10:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"727568fc0eaf6b87a6d0ec2bd58682cf01db1e6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 16:11:22.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad5bc48653b046eeb15560c34e738855 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/933105802f604dd0bb002fce66c5baac : SUCCESS in 4m 29s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/5e2acc0ca74f48ec873deac35f5272fe : SUCCESS in 40m 20s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/4dd9f9608a4040cbb68af8f478955a7b : FAILURE in 43m 33s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/2baa3ceeac5049489c9e62dca0d6ec2a : SUCCESS in 40m 00s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/5eaa1ee42d7742aca350b61a04bb9aff : SUCCESS in 57m 35s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/347dc6059ac644d1bc112f015f294252 : SUCCESS in 33m 26s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/28449566cecd4b7888fa6b75d46f739a : SUCCESS in 34m 45s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/adbdc46747d54b6d84df26f229736daa : SUCCESS in 36m 49s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/8b3b0118aca6429090fbb59b7150302b : SUCCESS in 45m 25s","accounts_in_message":[],"_revision_number":2},{"id":"bb3363916835129534e5ce93bfca0e9822544907","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2022-04-08 16:33:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"18fabe188c9ea2f1f8ac5000d49c5fa2e7ab6c35","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-04-08 17:28:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b5a5873adcbf333c50a7687d6c9baafe62d4fdfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 17:58:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efdf5eeb5ffd4996897667e8c01921a5 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46250f8180a842e6bd2e552fea664dc9 : SUCCESS in 5m 29s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/0f38790fb4ea48e39272c36dd7cc09b6 : SUCCESS in 57m 38s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/8a995a43cc7c4bbaab93117303623733 : SUCCESS in 41m 52s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/5e963beff14d486aa2ea5a0d5cee9a18 : SUCCESS in 1h 03m 09s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/2455d71dc2164e36929e31136dd8bd6b : SUCCESS in 51m 21s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/cbcf340922a748ef85f6514205fb4b40 : SUCCESS in 45m 37s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/bf013c7edfb240bf85520e4c9db0facf : SUCCESS in 1h 15m 49s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/e9b6c50d3cd84925917333ee4dd898de : SUCCESS in 59m 08s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9cdefb9be6a34a19916f2c7d728878f2 : SUCCESS in 45m 41s","accounts_in_message":[],"_revision_number":2},{"id":"91ef220be1081bea81c59e16bcc173d74bd97d68","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-04-08 18:05:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1423fe9b155c99a3351673d3cdcf103e270b2dc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 18:06:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5a87469d46d544644784ee2ec7e7408944361e80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 19:17:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb796b7540054b72a6b15578dd87567d : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5c9bc12a1154295b152d7a3ac702b97 : SUCCESS in 4m 05s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/2688431ff206401bbc8e55a4a6356468 : SUCCESS in 43m 27s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/37122e6d08634b2f8c8dcbf485db9ec1 : SUCCESS in 58m 35s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a7279b385edd49b4b4a3cdc20194905a : SUCCESS in 47m 05s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/2b6bdc806aee4ad08c4e8cee573226f8 : SUCCESS in 1h 06m 09s","accounts_in_message":[],"_revision_number":2},{"id":"565e0f1e4d3abe7576438e3701eba64a49dff341","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 19:17:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ecb6e851a81cd160ed58b47676ced899ffdc71dc","revisions":{"a21220789f69053063047cf20e3a287ba0071d5b":{"kind":"REWORK","_number":1,"created":"2022-04-08 12:44:10.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/21/837121/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/21/837121/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/21/837121/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/21/837121/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/21/837121/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/21/837121/1"}}},"commit":{"parents":[{"commit":"f8636615169a650ae756bad99378f235e5c82ade","subject":"Cinder: Remove tenant_id templating from endpoint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f8636615169a650ae756bad99378f235e5c82ade"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2022-04-08 12:43:37.000000000","tz":120},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2022-04-08 12:43:37.000000000","tz":120},"subject":"Support CS9 when installing RDO repo","message":"Support CS9 when installing RDO repo\n\nChange-Id: I5618353cb368e9d56188297dfdf87b1b127ee653\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a21220789f69053063047cf20e3a287ba0071d5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a21220789f69053063047cf20e3a287ba0071d5b"}]},"branch":"refs/heads/master"},"ecb6e851a81cd160ed58b47676ced899ffdc71dc":{"kind":"REWORK","_number":2,"created":"2022-04-08 13:19:05.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/21/837121/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/21/837121/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/21/837121/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/21/837121/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/21/837121/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/21/837121/2"}}},"commit":{"parents":[{"commit":"f8636615169a650ae756bad99378f235e5c82ade","subject":"Cinder: Remove tenant_id templating from endpoint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f8636615169a650ae756bad99378f235e5c82ade"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2022-04-08 12:43:37.000000000","tz":120},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2022-04-08 13:18:59.000000000","tz":120},"subject":"Support CS9 when installing RDO repo","message":"Support CS9 when installing RDO repo\n\nChange-Id: I5618353cb368e9d56188297dfdf87b1b127ee653\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ecb6e851a81cd160ed58b47676ced899ffdc71dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ecb6e851a81cd160ed58b47676ced899ffdc71dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
