)]}'
{"id":"x%2Fpackstack~837694","triplet_id":"x%2Fpackstack~master~I5764e11a39cacef6951c9fe57bb4b1bd91a74b7d","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-10-06 12:07:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5764e11a39cacef6951c9fe57bb4b1bd91a74b7d","subject":"Remove puppet-certmonger from dependencies","status":"MERGED","created":"2022-04-13 07:53:48.000000000","updated":"2022-10-06 12:07:02.000000000","submitted":"2022-10-06 12:07:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837694","meta_rev_id":"32e5fea42c49a038979f4943d8f17668b996acb4","_number":837694,"virtual_id_number":837694,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-06 12:07:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2022-10-06 11:18:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2022-10-06 11:18:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-13 08:57:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-06 11:18:36.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"a3093a467cbf1cf9aa5646c34e54964c53ef6260","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-13 07:53:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b551d7872554bed754a30d687feceb29dd90ce56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 08:57:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba13e13809f044ea87c0c6e5699be90a : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40aff9c17195496aaefa83b71ff12f5f : SUCCESS in 3m 59s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/64e4754e7677439394fe5f5ed8319aed : SUCCESS in 38m 39s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/32f3138914fc4a47b6e3f69290c4e154 : SUCCESS in 53m 05s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/347316a1e8a04be68365555b68582098 : SUCCESS in 42m 43s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/0dc3aa4f1702405fb742875e1b55d854 : SUCCESS in 47m 29s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/1889f388134a4f749b9a7af237a498ea : SUCCESS in 45m 45s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/7f211d0dbe23445290339ae2b985aedf : SUCCESS in 46m 27s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/87530d011bea400d9609d2b1505a2762 : SUCCESS in 37m 29s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/5d928e2b508a4b67b01324bcce461da5 : SUCCESS in 57m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e1ebf9cb0298c02a3847cf5cf8d8881ff84ad4a5","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-10-06 11:18:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"62d25272b13d38eb1f271fa026dd216061b91a59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 11:19:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70e16330a6c2b3f6762682a29c7a6f73bd811c7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 12:07:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/025df2d387ec4376b7e41024e4f153a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e42392ffe1f4833bdb6d733f5f5e064 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29de8dee923b408d8456de0625f45dcb : SUCCESS in 4m 18s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/4a2aecd0d83941f6adb8bff218695c04 : SUCCESS in 38m 09s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/6f503b5165a34d5e85a5c5dd128b3c55 : SUCCESS in 39m 09s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/c7816ebea109425b8e343d21bc0261c5 : SUCCESS in 41m 03s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ff77cbe490204f3db8442ef14f57f155 : SUCCESS in 45m 24s","accounts_in_message":[],"_revision_number":1},{"id":"32e5fea42c49a038979f4943d8f17668b996acb4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 12:07:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"95e812dfec58f0b4e4dca935524e24ae5385b431","revisions":{"95e812dfec58f0b4e4dca935524e24ae5385b431":{"kind":"REWORK","_number":1,"created":"2022-04-13 07:53:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/837694/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/94/837694/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/94/837694/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/94/837694/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/94/837694/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/94/837694/1"}}},"commit":{"parents":[{"commit":"ecb6e851a81cd160ed58b47676ced899ffdc71dc","subject":"Support CS9 when installing RDO repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ecb6e851a81cd160ed58b47676ced899ffdc71dc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-13 07:53:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-13 07:53:19.000000000","tz":540},"subject":"Remove puppet-certmonger from dependencies","message":"Remove puppet-certmonger from dependencies\n\n... because it is not used in deployment process.\n\nChange-Id: I5764e11a39cacef6951c9fe57bb4b1bd91a74b7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/95e812dfec58f0b4e4dca935524e24ae5385b431"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/95e812dfec58f0b4e4dca935524e24ae5385b431"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[]}
