)]}'
{"id":"x%2Fpackstack~842716","triplet_id":"x%2Fpackstack~master~Ic627bfa002c190706c49cd3ffe7f7b7a8cf1c27f","project":"x/packstack","branch":"master","topic":"trove-cleanup","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-06-09 12:28:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic627bfa002c190706c49cd3ffe7f7b7a8cf1c27f","subject":"Trove: Remove trove::use_neutron","status":"MERGED","created":"2022-05-20 13:46:45.000000000","updated":"2022-06-09 12:28:53.000000000","submitted":"2022-06-09 12:28:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842716-trove-cleanup","meta_rev_id":"3a5cdac4149d879f31a96322737f29c72773c1c3","_number":842716,"virtual_id_number":842716,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-09 12:28:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2022-06-09 11:36:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2022-06-09 11:36:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-20 15:02:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-09 11:36:36.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"a5e10ba85b3d2543ec6f502a74ecbbdbdf5492a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-20 13:46:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebc3a948bad81611c14ca7671ef6de8d1b8614db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-20 13:49:38.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6d8a408a38360fc1e12be521035f43d9ebe5f775","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 15:02:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20bcc6f81f864a82aa74cf2aefdfbe77 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0eddd43abbd149d2a05bf492f86e9b2f : SUCCESS in 3m 14s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/39a414c42aa04d17b9aa15506ca8ee79 : SUCCESS in 35m 12s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/b8f6cbc2ffaa43df9cad37cbd9680dc3 : SUCCESS in 32m 31s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/67ad1d51a25846f0b185057fcf01d07f : SUCCESS in 46m 27s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/48c129daf69049698f1a255e3efec9da : SUCCESS in 1h 05m 17s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/acd5d59b8146461c990c12f7671c53eb : SUCCESS in 35m 11s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/c1218a6b4190452bb44331892f44e2c5 : SUCCESS in 39m 19s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/8a76cd70a40a43eb94da93e88bfe7fee : SUCCESS in 32m 48s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d04f5b751222476192125a205c241f7d : SUCCESS in 40m 01s","accounts_in_message":[],"_revision_number":2},{"id":"42d38500f34489745b330da726d5b89bdcd1d976","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-25 12:07:16.000000000","message":"Topic set to trove-cleanup","accounts_in_message":[],"_revision_number":2},{"id":"aa1f3dcb227de50aee2e61717dcb8d7eed707e72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-06-08 13:54:10.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"19392c886b254e10b6c06dbfe9f1ade98dabb05c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 14:53:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beb4a4ffee0d44c6b7c1f46660a4eeb3 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91ad86948e9140ea80865de3c8587d70 : SUCCESS in 3m 53s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/10823efd71334dedac06fa03abffbb58 : SUCCESS in 51m 47s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d5d7b0bf31c640418c360f9a9a902fbe : SUCCESS in 45m 05s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/6a3bf8d3a85543448ede75bb81c25a48 : SUCCESS in 54m 55s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/29ca01e978294624836ac07c1082c373 : SUCCESS in 48m 19s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/5528d51562784d6dac327d6606a7f63e : SUCCESS in 41m 11s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/cb81bb8b383f4a76979310262aa43db1 : SUCCESS in 40m 30s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/e7162593fa884ebca7faa4471df523fd : SUCCESS in 40m 33s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/12d8609f4fb04607be4fc931c343bc88 : SUCCESS in 45m 26s","accounts_in_message":[],"_revision_number":3},{"id":"dc3934ec15a471f6b0c3949e628afe3260795215","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-06-09 11:36:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1e37b430e4b54f249bde9d6ca996bbc5296a2ced","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 11:37:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5875405ae4445ca2f4016f2a5dba26da5b0479bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 12:28:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4f2bb17b556436c8c02d383ca48723f : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d934af09688458fb5de93c4d27a27cd : SUCCESS in 3m 40s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/2d88bca2927546438668224a76a1b7d9 : SUCCESS in 41m 19s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/3ba516142ea94e4b90b42bfeb1e1e747 : SUCCESS in 43m 51s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/dae353309ab444daa6316d885b3d1969 : SUCCESS in 48m 21s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f918bcb2381041a0b21afdecaf772481 : SUCCESS in 50m 41s","accounts_in_message":[],"_revision_number":3},{"id":"3a5cdac4149d879f31a96322737f29c72773c1c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 12:28:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"edcd363dc6a8850fb417239bcd4986d7e55e8c0f","revisions":{"eb4eab7309ecb94bddb0d1fbe2d447e1b2be9e2e":{"kind":"REWORK","_number":1,"created":"2022-05-20 13:46:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/842716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/16/842716/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/16/842716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/16/842716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/16/842716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/16/842716/1"}}},"commit":{"parents":[{"commit":"f8636615169a650ae756bad99378f235e5c82ade","subject":"Cinder: Remove tenant_id templating from endpoint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f8636615169a650ae756bad99378f235e5c82ade"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-20 13:44:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-20 13:46:33.000000000","tz":540},"subject":"Trove: Remove trove::use_neutron","message":"Trove: Remove trove::use_neutron\n\n... because the parameter was deprecated during Yoga cycle[1] and has\nhad no effect since then.\n\n[1] https://review.opendev.org/c/openstack/puppet-trove/+/830620\n\nChange-Id: Ic627bfa002c190706c49cd3ffe7f7b7a8cf1c27f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/eb4eab7309ecb94bddb0d1fbe2d447e1b2be9e2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/eb4eab7309ecb94bddb0d1fbe2d447e1b2be9e2e"}]},"branch":"refs/heads/master"},"b006d6e3ccd61c3b991824e8d60c582447f44fe7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-05-20 13:49:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/842716/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/16/842716/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/16/842716/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/16/842716/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/16/842716/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/16/842716/2"}}},"commit":{"parents":[{"commit":"f8636615169a650ae756bad99378f235e5c82ade","subject":"Cinder: Remove tenant_id templating from endpoint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f8636615169a650ae756bad99378f235e5c82ade"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-20 13:44:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-20 13:49:21.000000000","tz":540},"subject":"Trove: Remove trove::use_neutron","message":"Trove: Remove trove::use_neutron\n\n... because the parameter was deprecated during Yoga cycle[1] and has\nhad no effect since then.\n\nDepends-on: https://review.opendev.org/c/openstack/puppet-trove/+/830620\nChange-Id: Ic627bfa002c190706c49cd3ffe7f7b7a8cf1c27f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b006d6e3ccd61c3b991824e8d60c582447f44fe7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b006d6e3ccd61c3b991824e8d60c582447f44fe7"}]},"branch":"refs/heads/master"},"edcd363dc6a8850fb417239bcd4986d7e55e8c0f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-06-08 13:54:10.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/16/842716/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/16/842716/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/16/842716/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/16/842716/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/16/842716/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/16/842716/3"}}},"commit":{"parents":[{"commit":"552b45be5e3646c8d6809f890ddb89da015ec632","subject":"Fix dstat options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/552b45be5e3646c8d6809f890ddb89da015ec632"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-20 13:44:51.000000000","tz":540},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-06-08 13:54:10.000000000","tz":0},"subject":"Trove: Remove trove::use_neutron","message":"Trove: Remove trove::use_neutron\n\n... because the parameter was deprecated during Yoga cycle[1] and has\nhad no effect since then.\n\nDepends-on: https://review.opendev.org/c/openstack/puppet-trove/+/830620\nChange-Id: Ic627bfa002c190706c49cd3ffe7f7b7a8cf1c27f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/edcd363dc6a8850fb417239bcd4986d7e55e8c0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/edcd363dc6a8850fb417239bcd4986d7e55e8c0f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
