)]}'
{"id":"x%2Fpackstack~845014","triplet_id":"x%2Fpackstack~master~I0a77a4711a8eb03deda15574ba309ae2df5ccbeb","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-06-08 07:28:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0a77a4711a8eb03deda15574ba309ae2df5ccbeb","subject":"Fix dstat options","status":"MERGED","created":"2022-06-07 16:24:35.000000000","updated":"2022-06-08 07:28:01.000000000","submitted":"2022-06-08 07:28:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845014","meta_rev_id":"a1cf37864ec1cfdc8cd2763c15f47db1de8757b2","_number":845014,"virtual_id_number":845014,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-08 07:28:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2022-06-08 06:17:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2022-06-08 06:17:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-07 17:36:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-08 06:17:24.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"b30b8661a28d07ad70db6d775e854518cc025931","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-07 16:24:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f553f60e40586d08983c34ed44aee8e15e60355","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-07 17:36:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60cce62c3da3433aa14b9d22b7cb9a6e : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eacf9ef8e24244ebb7bcb90cf00c6f53 : SUCCESS in 3m 49s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/6d1076afd82744c89613a67c09c319c2 : SUCCESS in 1h 02m 12s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/b24baf031f85410593c52a8e36269846 : SUCCESS in 58m 06s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/f7c4c225a767439c8720d6abea452567 : SUCCESS in 1h 02m 19s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/47ad4eff993d4b679b3985ea795cb11f : SUCCESS in 1h 05m 04s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/081225ac638446fe864ab71bd0d698dd : SUCCESS in 54m 51s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/098cf2aff5414c8b8cf305f765817ac9 : SUCCESS in 52m 21s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/bab822d2d690457b88a901a8c5744d91 : SUCCESS in 50m 55s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/cb2e82eedd2a4735901a9901d77271c6 : SUCCESS in 57m 44s","accounts_in_message":[],"_revision_number":1},{"id":"cf581cb614108c7d80054061a0fdd7700d4c3e9f","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-06-08 06:17:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3a8d472b1f2e3a236dad0480020dca4637a2bc76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 06:18:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dc0b21a22485809fa992351a9b242ad099f7766e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 07:28:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b22e7cdec897496b99a0baf5e4bb1e8e : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad8ad06097e0434ba7cd1ca6a5bcb174 : SUCCESS in 3m 04s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/da19723d623e4390b3303feacd98fdf6 : SUCCESS in 37m 42s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/2f8db5b9c54440cfbe804ca15a579c65 : SUCCESS in 36m 21s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/e0c6e92e746a497ea797fb7ed2357b4b : SUCCESS in 42m 39s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/3ed40c730f024d7381f267339ec5864b : SUCCESS in 45m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a1cf37864ec1cfdc8cd2763c15f47db1de8757b2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 07:28:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"552b45be5e3646c8d6809f890ddb89da015ec632","revisions":{"552b45be5e3646c8d6809f890ddb89da015ec632":{"kind":"REWORK","_number":1,"created":"2022-06-07 16:24:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/845014/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/14/845014/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/14/845014/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/14/845014/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/14/845014/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/14/845014/1"}}},"commit":{"parents":[{"commit":"d61fd5f2280db154bac03637c95a887f07808568","subject":"Add required fixes to pass CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d61fd5f2280db154bac03637c95a887f07808568"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-07 16:22:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-07 16:22:51.000000000","tz":540},"subject":"Fix dstat options","message":"Fix dstat options\n\n--top-io-adv and --top-cpu-adv aren\u0027t options in c8, so let\u0027s make\ntheir usage dynamic.\n\nThis imports https://review.opendev.org/801659 from the upstream\np-o-i repo.\n\nCo-Authored-By: Alex Schultz \u003caschultz@next-development.com\u003e\nChange-Id: I0a77a4711a8eb03deda15574ba309ae2df5ccbeb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/552b45be5e3646c8d6809f890ddb89da015ec632"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/552b45be5e3646c8d6809f890ddb89da015ec632"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
