)]}'
{"id":"x%2Fpackstack~848487","triplet_id":"x%2Fpackstack~master~Ic9930107fbc68cba3432f4424f113071325efcb7","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-10-06 10:55:58.000000000","reason":"Change was submitted"},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2022-10-06 08:40:33.000000000","reason":"\u003cGERRIT_ACCOUNT_16312\u003e replied on the change","reason_account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}},"hashtags":[],"change_id":"Ic9930107fbc68cba3432f4424f113071325efcb7","subject":"Replace hiera by lookup","status":"MERGED","created":"2022-07-02 14:52:33.000000000","updated":"2022-10-06 10:55:58.000000000","submitted":"2022-10-06 10:55:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848487","meta_rev_id":"90217727557cfbbb1ff30c9c5f3bf6af3dd9fd2d","_number":848487,"virtual_id_number":848487,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-06 10:55:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2022-10-06 08:40:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2022-10-06 08:47:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-02 15:10:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-16 10:18:33.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"CC"},{"updated":"2022-10-06 08:40:33.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"fc3b65692f981e31e18af9a74577bc88d44196a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-02 14:52:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f4aaee26d103dd6a7aa0935e99ccda4cdec8bde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-02 15:10:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7bcbd30cf5842dfac8e01486536be1c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c84ef07daa14eff9e1d154ca2634065 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8db8f5bea0fc4ecfb5ac84c07c3d9f11 : SUCCESS in 3m 44s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/c4017a0a9c1144d998ace06a1cb2fc62 : FAILURE in 10m 36s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/8d89029c6ed14681a31daf28dd0ae495 : FAILURE in 11m 14s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/e3c5532f0cf94644b1a215c74df64a0c : FAILURE in 10m 06s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f425d88a623c4426a8995990b5516adc : FAILURE in 11m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d41b80d1b65245832abd4877c2b5657640a762ab","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-08-16 10:18:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"501933e91368fa25d8c2a12b82cf23271b29718e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 10:43:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf84b1af5ac340bc8bda2d9b59588651\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f41a1f0c8c264fc0b09a17ccaff9b8c5 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f876d063ce54ebab0df103ca7ca6025 : SUCCESS in 4m 29s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/b776094170fa4d5cb8d9452c62b9f538 : FAILURE in 15m 29s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/67bdb67427854662820d3417f56a42cf : FAILURE in 15m 16s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/5e11a1617d414428af4ee181b5b43e5b : FAILURE in 19m 07s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/7c86368e6d654eb3a2f69c79139f549b : FAILURE in 15m 21s","accounts_in_message":[],"_revision_number":1},{"id":"99e60eb0052995dc8ed7630a7350f2745dcf6c62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-16 11:34:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"827f2b067536b23fcfaf8c017e28cb889f5eb6eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 12:05:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f76b7061d264ebb93d242ebd03d907f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e42f2ca801874793922c15d15a9187f4 : SUCCESS in 7m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb1fa7dabf3c4916bf6883285faa60fc : SUCCESS in 3m 52s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/e5b4ca50522c491eb1be53b66d3bdc33 : FAILURE in 19m 56s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/8df76d7a4f374717abe12bfbccd11016 : FAILURE in 19m 22s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/27a49fc3f63a4b6e85cb6c2cd0e24d9c : FAILURE in 19m 39s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ff8efad52d7c42898b9a5d1c2e078d82 : FAILURE in 22m 48s","accounts_in_message":[],"_revision_number":2},{"id":"a1d4d20dbd0f9e0dc738723ab9855d2365751d75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-16 12:34:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8bf3adb62710af1d7eef142d68a3921c53d6719e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 13:25:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e068bc4416114ff1a90c17e426217357\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/852207611ac84b77aa8ceb474dce0bb5 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51eff4ecc46f406887bd689d2a763711 : SUCCESS in 4m 33s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/5835852cfcc34284aee9dd3a2272ca69 : SUCCESS in 39m 20s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/67901a589181457d9c97f0c6c0deedc6 : SUCCESS in 39m 59s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/3cbf1a51c3ae4e4dbb4283bb73a9a183 : SUCCESS in 43m 16s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/b4ac0286eb62486e8da93a07f04acb8c : SUCCESS in 48m 37s","accounts_in_message":[],"_revision_number":3},{"id":"49393c032775251a8f5be68fc46e7b103e12fb85","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-08-17 09:32:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"73b83b117a5240bd5c8e6f05c3d4445ced2db1e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-17 13:49:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4f7c15361e01340ee70a720aca591d6c22da6d24","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-17 13:50:03.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b994d15c92492d16587436ae4897c4b2870d2f67","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-08-17 13:55:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bea2f32e51433305ad9df5ef7abce5bb69b9e85f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 14:27:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f8028efba634797b1528642b421ec89\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e5fe413b25342de8affdddc8d9f3f2e : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8c7f30ea0354f0cbd21d1fc239f664f : SUCCESS in 3m 49s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/0e72f11410724c5f98b2f36d9a28b579 : FAILURE in 23m 36s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/7d8b983c12d94944a1111d97087f8d91 : FAILURE in 25m 06s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ad7c65a1c9b645d89af0a25c578f8750 : FAILURE in 24m 22s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/129207a6dc754d4d88ef7009a13d5cae : FAILURE in 27m 31s","accounts_in_message":[],"_revision_number":4},{"id":"248a310cefb47ff155f2029f261439918004b940","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-17 15:38:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a8392415e9189d52d679fc6560a15f6b1c170c80","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-17 15:45:12.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"06fa8f9eebedc675b81b3d3fce60e16b89519cb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 16:35:21.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/75f26a9b887f4bdfbf7a086bb1ba5ac9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbb12e445a2a40039d19b93ee796355d : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4bb9fd75763143d4909791dc826cecec : SUCCESS in 4m 15s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/4a617a4503054afca9f8765a0ec855dc : SUCCESS in 44m 53s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/3e4f4f2e9c6840378f5a6f4c41e1ff52 : SUCCESS in 42m 36s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/6fa204ed072a47569676eef241135b1a : SUCCESS in 47m 37s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/4b0c13fa1afe4c6ba58acae93b05d73b : FAILURE in 54m 07s","accounts_in_message":[],"_revision_number":5},{"id":"1553a6d161950fc55c7e518ce9372e1c7e0d58a1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-20 18:04:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2ca6baf6157b0c282d25e20a85e4a58aa94e6d2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-20 19:11:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ffc7ed2cc1c486b8b1232ad5b67781b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdf1ecc0443d42e8a9f3fead1356d863 : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5f2707bc4d94fcabae4c203dcf7511d : SUCCESS in 4m 02s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/94b024086e8a4e77878bb822c65e5a08 : SUCCESS in 36m 18s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/c6359763b79742de9a77db04c07ef76f : SUCCESS in 36m 26s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/c82afde669234a89879c8a2d2b153132 : SUCCESS in 36m 15s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/cc6477f622b64d8fb77d0f01ff1431ae : SUCCESS in 1h 01m 52s","accounts_in_message":[],"_revision_number":5},{"id":"ccaefda7181261ebd692fbee6ccc0533d9f6a00e","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-10-06 08:40:33.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"90b7267721ad60a68809654fcea31dc8ad8592cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 08:41:22.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d680bdb95bb710e87469d591b480b3623f27bec6","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-10-06 08:41:54.000000000","message":"Patch Set 5: -Workflow","accounts_in_message":[],"_revision_number":5},{"id":"07f96515aa369d3394594b0e118042ee146b94fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 08:42:06.000000000","message":"Patch Set 5: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":5},{"id":"43e56f29d00f9db08dcfc42969bd4086fd5ac94a","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-10-06 08:47:18.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"294a286be6f6e92dc4cb06bc113d7a90bb23f228","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 09:47:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5465f017002b474a83a750a8fcd2ecb9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7edc913ed78496da9071737e7b5584c : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25f4414fdc734384adb7046b8faee91b : SUCCESS in 3m 53s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/6e6160eef89049eaab51b82ea5a49685 : SUCCESS in 53m 21s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/fb2554977b7546ccb0a81d167ab6cba2 : SUCCESS in 50m 09s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/0e83363166154dfebc9eeec62d726b49 : SUCCESS in 53m 54s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/693639a772644173afeaacffa636e33c : SUCCESS in 54m 04s","accounts_in_message":[],"_revision_number":5},{"id":"f23b9634924a8c1c4ee7bafc18d71118ebffd82c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 09:47:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"209fac0b1f17d68431fff40b3019ea3418ce90b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 10:55:58.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48d03c190764464eba95610e2503a7c8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1b3de18f5384de597cc63936a20a927 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f0724bbd71d41d1856e967e05c8dd5e : SUCCESS in 5m 16s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/f27132e31fd3445fa3bfbef06c535c29 : SUCCESS in 41m 26s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d6b1cff6933b4338ad6e5d8c12e95c76 : SUCCESS in 56m 26s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/83afc752e0684f13940cc722c55a8929 : SUCCESS in 43m 44s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ef8ac51d72d94f6e8c1f2a68d83b6b2b : SUCCESS in 1h 03m 39s","accounts_in_message":[],"_revision_number":5},{"id":"90217727557cfbbb1ff30c9c5f3bf6af3dd9fd2d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 10:55:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6ff87dd6d4f4d9031a10be9544d24c20fd34244e","revisions":{"9d8f2ef566e309471112ae02b52632eba4e79b92":{"kind":"REWORK","_number":1,"created":"2022-07-02 14:52:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/848487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/87/848487/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/87/848487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/87/848487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/87/848487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/87/848487/1"}}},"commit":{"parents":[{"commit":"6f9aedc4a3f95b0b14af71716e8433fee09e0621","subject":"Trove: Use trove::db to set up database options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6f9aedc4a3f95b0b14af71716e8433fee09e0621"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-02 14:50:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-02 14:50:33.000000000","tz":540},"subject":"Replace hiera by lookup","message":"Replace hiera by lookup\n\nThe hiera function is deprecated and does not work with the latest\nhieradata version 5. It should be replaced by the new lookup\nfunction[1].\n\n[1] https://puppet.com/docs/puppet/7/hiera_automatic.html\n\nWith the lookup function, we can define value type and merge behavior,\nbut these are kept default at this moment to limit scope of this change\nto just simple replacement. Adding value type might be useful to make\nsure the value is in expected type (especially when a boolean value is\nexpected), but we will revisit that later.\n\nexample:\nlookup(\u003cNAME\u003e, [\u003cVALUE TYPE\u003e], [\u003cMERGE BEHAVIOR\u003e], [\u003cDEFAULT VALUE\u003e])\n\nChange-Id: Ic9930107fbc68cba3432f4424f113071325efcb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d8f2ef566e309471112ae02b52632eba4e79b92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d8f2ef566e309471112ae02b52632eba4e79b92"}]},"branch":"refs/heads/master"},"db99f8792d3019fa330444923c1c2457769ddb98":{"kind":"REWORK","_number":2,"created":"2022-08-16 11:34:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/848487/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/87/848487/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/87/848487/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/87/848487/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/87/848487/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/87/848487/2"}}},"commit":{"parents":[{"commit":"6f9aedc4a3f95b0b14af71716e8433fee09e0621","subject":"Trove: Use trove::db to set up database options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6f9aedc4a3f95b0b14af71716e8433fee09e0621"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-02 14:50:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-16 11:33:42.000000000","tz":540},"subject":"Replace hiera by lookup","message":"Replace hiera by lookup\n\nThe hiera function is deprecated and does not work with the latest\nhieradata version 5. It should be replaced by the new lookup\nfunction[1].\n\n[1] https://puppet.com/docs/puppet/7/hiera_automatic.html\n\nWith the lookup function, we can define value type and merge behavior,\nbut these are kept default at this moment to limit scope of this change\nto just simple replacement. Adding value type might be useful to make\nsure the value is in expected type (especially when a boolean value is\nexpected), but we will revisit that later.\n\nexample:\nlookup(\u003cNAME\u003e, [\u003cVALUE TYPE\u003e], [\u003cMERGE BEHAVIOR\u003e], [\u003cDEFAULT VALUE\u003e])\n\nChange-Id: Ic9930107fbc68cba3432f4424f113071325efcb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/db99f8792d3019fa330444923c1c2457769ddb98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/db99f8792d3019fa330444923c1c2457769ddb98"}]},"branch":"refs/heads/master"},"52518ae46b8274d0a6fd035af51c776a2a84f081":{"kind":"REWORK","_number":3,"created":"2022-08-16 12:34:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/848487/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/87/848487/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/87/848487/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/87/848487/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/87/848487/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/87/848487/3"}}},"commit":{"parents":[{"commit":"6f9aedc4a3f95b0b14af71716e8433fee09e0621","subject":"Trove: Use trove::db to set up database options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6f9aedc4a3f95b0b14af71716e8433fee09e0621"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-02 14:50:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-16 12:34:19.000000000","tz":540},"subject":"Replace hiera by lookup","message":"Replace hiera by lookup\n\nThe hiera function is deprecated and does not work with the latest\nhieradata version 5. It should be replaced by the new lookup\nfunction[1].\n\n[1] https://puppet.com/docs/puppet/7/hiera_automatic.html\n\nWith the lookup function, we can define value type and merge behavior,\nbut these are kept default at this moment to limit scope of this change\nto just simple replacement. Adding value type might be useful to make\nsure the value is in expected type (especially when a boolean value is\nexpected), but we will revisit that later.\n\nexample:\nlookup(\u003cNAME\u003e, [\u003cVALUE TYPE\u003e], [\u003cMERGE BEHAVIOR\u003e], [\u003cDEFAULT VALUE\u003e])\n\nChange-Id: Ic9930107fbc68cba3432f4424f113071325efcb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/52518ae46b8274d0a6fd035af51c776a2a84f081"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/52518ae46b8274d0a6fd035af51c776a2a84f081"}]},"branch":"refs/heads/master"},"fa63bb8d02191b7abefdbaa5c972ef666ef567d9":{"kind":"REWORK","_number":4,"created":"2022-08-17 13:49:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/848487/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/87/848487/4","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/87/848487/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/87/848487/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/87/848487/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/87/848487/4"}}},"commit":{"parents":[{"commit":"6f9aedc4a3f95b0b14af71716e8433fee09e0621","subject":"Trove: Use trove::db to set up database options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6f9aedc4a3f95b0b14af71716e8433fee09e0621"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-02 14:50:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-17 13:48:53.000000000","tz":540},"subject":"Replace hiera by lookup","message":"Replace hiera by lookup\n\nThe hiera function is deprecated and does not work with the latest\nhieradata version 5. It should be replaced by the new lookup\nfunction[1].\n\n[1] https://puppet.com/docs/puppet/7/hiera_automatic.html\n\nWith the lookup function, we can define value type and merge behavior,\nbut these are kept default at this moment to limit scope of this change\nto just simple replacement. Adding value type might be useful to make\nsure the value is in expected type (especially when a boolean value is\nexpected), but we will revisit that later.\n\nexample:\nlookup(\u003cNAME\u003e, [\u003cVALUE TYPE\u003e], [\u003cMERGE BEHAVIOR\u003e], [\u003cDEFAULT VALUE\u003e])\n\nChange-Id: Ic9930107fbc68cba3432f4424f113071325efcb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fa63bb8d02191b7abefdbaa5c972ef666ef567d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fa63bb8d02191b7abefdbaa5c972ef666ef567d9"}]},"branch":"refs/heads/master"},"6ff87dd6d4f4d9031a10be9544d24c20fd34244e":{"kind":"REWORK","_number":5,"created":"2022-08-17 15:38:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/848487/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/87/848487/5","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/87/848487/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/87/848487/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/87/848487/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/87/848487/5"}}},"commit":{"parents":[{"commit":"6f9aedc4a3f95b0b14af71716e8433fee09e0621","subject":"Trove: Use trove::db to set up database options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6f9aedc4a3f95b0b14af71716e8433fee09e0621"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-02 14:50:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-17 15:37:32.000000000","tz":540},"subject":"Replace hiera by lookup","message":"Replace hiera by lookup\n\nThe hiera function is deprecated and does not work with the latest\nhieradata version 5. It should be replaced by the new lookup\nfunction[1].\n\n[1] https://puppet.com/docs/puppet/7/hiera_automatic.html\n\nWith the lookup function, we can define value type and merge behavior,\nbut these are kept default at this moment to limit scope of this change\nto just simple replacement. Adding value type might be useful to make\nsure the value is in expected type (especially when a boolean value is\nexpected), but we will revisit that later.\n\nexample:\nlookup(\u003cNAME\u003e, [\u003cVALUE TYPE\u003e], [\u003cMERGE BEHAVIOR\u003e], [\u003cDEFAULT VALUE\u003e])\n\nThis also replaces the hiera_array function which is also deprecated,\naccording to the guideline[2].\n\n[2] https://puppet.com/docs/puppet/7/hiera_migrate.html#updated_classic_hiera_function_calls\n\nChange-Id: Ic9930107fbc68cba3432f4424f113071325efcb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6ff87dd6d4f4d9031a10be9544d24c20fd34244e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6ff87dd6d4f4d9031a10be9544d24c20fd34244e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[]}
