)]}'
{"id":"x%2Fpackstack~852192","triplet_id":"x%2Fpackstack~master~I87929a7b94426f9a5d176131dc9a850d15c9084b","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-16 08:45:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I87929a7b94426f9a5d176131dc9a850d15c9084b","subject":"Mark Python 3.9 supported","status":"MERGED","created":"2022-08-04 16:30:48.000000000","updated":"2022-08-16 08:45:55.000000000","submitted":"2022-08-16 08:45:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852192","meta_rev_id":"eafaaff0f7b76f0a98a1a45908dce0ce89ac797d","_number":852192,"virtual_id_number":852192,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-16 08:45:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2022-08-16 07:54:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2022-08-16 07:54:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-04 17:10:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-16 07:54:09.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"b5c875ea5db67cfd563a11700121cea4e7f5eb49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-04 16:30:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b919ed9ce8565401116d0fe66fa8d0bc9f2b2520","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-04 16:31:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"77ec3e62f2a22c8e9383316815233de1cb45b27d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 17:10:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/62501b571b2548f2b4c09514c8389cbc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/58ce7705beb3487681e107dad0261fef : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e74b6854954d420aaa270891d4a76e7c : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e08fc3f5289480b88120fe813b5bd60 : SUCCESS in 6m 33s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/2bc5c1f9f53a4345b3f7bdf7464ebcb5 : FAILURE in 15m 08s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/5bba560d33b34d8da4df630efda453a5 : FAILURE in 20m 40s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/341c5a11f92d4093ba58c46a27114151 : FAILURE in 15m 23s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/acc1d88988674da88fe976a77beea02d : FAILURE in 23m 02s","accounts_in_message":[],"_revision_number":2},{"id":"044bed5f74724c9ba4f57f32953f9322b04f7eed","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 17:46:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3926d968e9b0ced33e0a51530abb70c697fddb30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 18:35:29.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ff01f8bd20d4d5993ae7abfeacf133c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/288fc7d372a7422e9d678899e888a206 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b24f6e99b1204c4e8abea9d8cb7f9d30 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35bc1598fd4a45789761cbf1bffb61a0 : SUCCESS in 6m 47s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/aa23bb4472cf4c6dae107fb07c7736fb : FAILURE in 20m 01s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/32ac6b1b330a4a10a23a2e966b61078c : FAILURE in 19m 38s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/473774c5b83a4310bc9705cf3604d39a : FAILURE in 19m 03s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a6417096b73245acab542038ff54370b : FAILURE in 22m 00s","accounts_in_message":[],"_revision_number":2},{"id":"3330ffc021a34906fa102dba29d40545f5616535","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-16 00:48:32.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"74e1607814051dc875fe937e1bc76c2ba0221286","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 01:51:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ba66ae80bd5445998c008ebf12ea9f2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f2d31d44c34e48429c181807486c5871 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c37f6fd5632b43879416b4b5d396970d : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7506371185743e3bee9fbe994dd2e30 : SUCCESS in 4m 35s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/11a6e4adcf1f4db5b7510a4e133ac776 : SUCCESS in 52m 08s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/e27fd2ce98304ab2be3ec59d53a27b57 : SUCCESS in 52m 33s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/9a8f7226138e4f89893fc4adeb7fb5f0 : SUCCESS in 53m 32s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/b1786d8dcccd466c84c9df4bb6cd1903 : SUCCESS in 58m 02s","accounts_in_message":[],"_revision_number":3},{"id":"e6216115de451436c507f82b4a3abfe44943c76a","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-08-16 07:54:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"de069033082f10cfca7741e6bcd2512b9ab995c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 07:54:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4bedb16f5352d6cc65e751bc2daac5d9d53ceee5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 08:45:54.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36f2eede0fad43ff9fc49f9df0311ea1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fb8efd62d5374d1d9556938a2065d958 : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d875d37f11342fe9622a69ba07bf02c : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f847d81dfa1e406ebf520e958d2dc9db : SUCCESS in 4m 11s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/58e8835ac0de4e3f8fca9be53cc4b2c4 : SUCCESS in 36m 41s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/30ee27276b414b268278b4c90e86f2ce : SUCCESS in 39m 43s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/7fa5d9583a8d4df4a242dcc12a35063e : SUCCESS in 40m 16s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/0ecb4dfd65774a76ad197ebf0b18ae33 : SUCCESS in 44m 42s","accounts_in_message":[],"_revision_number":3},{"id":"eafaaff0f7b76f0a98a1a45908dce0ce89ac797d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 08:45:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4378f0e5164510c9defb12b6c3b5c4ea32e83402","revisions":{"33003cd9724bcb70cb0272c2bfe2a1afcf9aba0f":{"kind":"REWORK","_number":1,"created":"2022-08-04 16:30:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/92/852192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/92/852192/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/92/852192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/92/852192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/92/852192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/92/852192/1"}}},"commit":{"parents":[{"commit":"6f9aedc4a3f95b0b14af71716e8433fee09e0621","subject":"Trove: Use trove::db to set up database options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6f9aedc4a3f95b0b14af71716e8433fee09e0621"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-04 16:29:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-04 16:29:35.000000000","tz":540},"subject":"Mark Python 3.9 supported","message":"Mark Python 3.9 supported\n\n... because CentOS Stream 9 has Python 3.9.\n\nChange-Id: I87929a7b94426f9a5d176131dc9a850d15c9084b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/33003cd9724bcb70cb0272c2bfe2a1afcf9aba0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/33003cd9724bcb70cb0272c2bfe2a1afcf9aba0f"}]},"branch":"refs/heads/master"},"5f3a4a6bd5c1a703c1b7992121e0bca6347079bf":{"kind":"REWORK","_number":2,"created":"2022-08-04 16:31:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/92/852192/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/92/852192/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/92/852192/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/92/852192/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/92/852192/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/92/852192/2"}}},"commit":{"parents":[{"commit":"6f9aedc4a3f95b0b14af71716e8433fee09e0621","subject":"Trove: Use trove::db to set up database options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6f9aedc4a3f95b0b14af71716e8433fee09e0621"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-04 16:29:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-04 16:31:00.000000000","tz":540},"subject":"Mark Python 3.9 supported","message":"Mark Python 3.9 supported\n\n... because CentOS Stream 9 has Python 3.9.\n\nChange-Id: I87929a7b94426f9a5d176131dc9a850d15c9084b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5f3a4a6bd5c1a703c1b7992121e0bca6347079bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5f3a4a6bd5c1a703c1b7992121e0bca6347079bf"}]},"branch":"refs/heads/master"},"4378f0e5164510c9defb12b6c3b5c4ea32e83402":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-08-16 00:48:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/92/852192/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/92/852192/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/92/852192/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/92/852192/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/92/852192/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/92/852192/3"}}},"commit":{"parents":[{"commit":"3d419162615c982c22040981deec3ea7973d8486","subject":"Stop overriding placement::wsgi::apache::api_port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3d419162615c982c22040981deec3ea7973d8486"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-04 16:29:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-16 00:48:32.000000000","tz":0},"subject":"Mark Python 3.9 supported","message":"Mark Python 3.9 supported\n\n... because CentOS Stream 9 has Python 3.9.\n\nChange-Id: I87929a7b94426f9a5d176131dc9a850d15c9084b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4378f0e5164510c9defb12b6c3b5c4ea32e83402"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4378f0e5164510c9defb12b6c3b5c4ea32e83402"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[]}
