)]}'
{"id":"x%2Fpackstack~852193","triplet_id":"x%2Fpackstack~master~Idb5fc37d9f334052839a0914895bf68f1b549037","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-16 08:55:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idb5fc37d9f334052839a0914895bf68f1b549037","subject":"Declare removal of Python 2 support","status":"MERGED","created":"2022-08-04 16:34:06.000000000","updated":"2022-08-16 08:55:04.000000000","submitted":"2022-08-16 08:55:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852193","meta_rev_id":"af1801970148126b5ae2ccfbd649e8db2f91cef3","_number":852193,"virtual_id_number":852193,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-16 08:55:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2022-08-16 07:54:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2022-08-16 07:54:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-04 16:54:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-16 07:54:29.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"fd20b94bda69350f9f14138cbefe8e9a6ec1593c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-04 16:34:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1732be77fe64cff7d9b767c6416af7eed62d8423","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 16:54:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4598712ec15c430995d4bc8693f1c5ef\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/90974399e4f34720967e7bb0dd93b45d : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d840924922744df9929f8e4466d9b26d : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61a063e78a334004b79da275bf635af8 : SUCCESS in 4m 49s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/670b249612d047529c6d8ffe774cdb06 : FAILURE in 12m 16s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/92a3a062cbd64bd7a9c920f6c4bf67d2 : FAILURE in 14m 33s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/abd65c21bbac438696e47497fa8d2e8d : FAILURE in 11m 55s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/541473aa25924fc6bf8345680b738bf7 : FAILURE in 17m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e4ac3d8a8811478d20ee0cfd7b3a73d77e6b21aa","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 17:47:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"baeb313ad905655e7e0a29c4ac76c43fac060e59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 18:06:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/020adebd13464eb5a6dbd84565246acb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/405a587d6c0d47d381d21d7720cafdd3 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/484a2fd9879a4475b640d7b25cf768c0 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/daca8ce1b2e1480d9e0e70289fba6077 : SUCCESS in 5m 11s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/333cf400667c47cb989faab2485c8bcc : FAILURE in 16m 01s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/054edf5bdef14052873ddd316384e1d7 : FAILURE in 15m 23s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/96bb509db59d439e860d27a3293908df : FAILURE in 16m 06s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/898ccf21495e4a0697edb54d79bca4a8 : FAILURE in 18m 00s","accounts_in_message":[],"_revision_number":1},{"id":"43cb978075ed755dc41c23cd0575e02ac92e6dc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-16 00:48:41.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"747ecdb96e5685ba61725b7d139ac866896e88de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 01:42:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8840189115b141ee9073d00d7f4687b2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0a8bbcc020494a5f9c34d5936cb18f54 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdfcedc16e154ac3bdfc15a4a139072a : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2864fa90c5ed490b89556d0ba77ed3f0 : SUCCESS in 7m 45s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/8086a2d63d15471681874b5279ca9ba0 : SUCCESS in 41m 12s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/158b037b817a47b0afc074a0ed180037 : SUCCESS in 43m 24s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/74950aeb8d374e47af24286a96bb10f2 : SUCCESS in 48m 09s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/b1309377010c4782ade41c553eb86d0e : SUCCESS in 52m 34s","accounts_in_message":[],"_revision_number":2},{"id":"ab9164609b8252413fcd0772d3352ee5637643d3","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-08-16 07:54:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8516b0064bcc396e0ba920729b3c39eefee0ddd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 07:55:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ba1f28b6591ab63eeb597a1ca05e6df178ee5b78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 08:55:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d90a58e862be4324a637ccf719eed339\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e85fa18f3ea541edbb9196128d294dc4 : SUCCESS in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bf1a96bda8c4fb39df957408397734d : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87e137fd90fd46e4a899891d183dfd29 : SUCCESS in 4m 30s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/ce0ff75e763344c9af5cf762bb81e715 : SUCCESS in 49m 04s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/2e5b99ef3d8d4c228f9b03be1e479430 : SUCCESS in 38m 09s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/1228fb32a2454923a4645be35f22a7f9 : SUCCESS in 55m 52s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/79db53feccd64c73b156ff0dc3abc759 : SUCCESS in 40m 33s","accounts_in_message":[],"_revision_number":2},{"id":"af1801970148126b5ae2ccfbd649e8db2f91cef3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 08:55:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"10460e0c14bcbee6a455ce14131e4965d3e25c95","revisions":{"a8a16a38d3af73aeacb9b5d3d1531891e1610cfd":{"kind":"REWORK","_number":1,"created":"2022-08-04 16:34:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/852193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/93/852193/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/93/852193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/93/852193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/93/852193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/93/852193/1"}}},"commit":{"parents":[{"commit":"5f3a4a6bd5c1a703c1b7992121e0bca6347079bf","subject":"Mark Python 3.9 supported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5f3a4a6bd5c1a703c1b7992121e0bca6347079bf"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-04 16:32:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-04 16:32:59.000000000","tz":540},"subject":"Declare removal of Python 2 support","message":"Declare removal of Python 2 support\n\n... because Python 2 reached to its EOL while ago and is no longer\navailable since CentOS 8.\n\nChange-Id: Idb5fc37d9f334052839a0914895bf68f1b549037\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a8a16a38d3af73aeacb9b5d3d1531891e1610cfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a8a16a38d3af73aeacb9b5d3d1531891e1610cfd"}]},"branch":"refs/heads/master"},"10460e0c14bcbee6a455ce14131e4965d3e25c95":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-08-16 00:48:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/852193/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/93/852193/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/93/852193/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/93/852193/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/93/852193/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/93/852193/2"}}},"commit":{"parents":[{"commit":"4378f0e5164510c9defb12b6c3b5c4ea32e83402","subject":"Mark Python 3.9 supported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4378f0e5164510c9defb12b6c3b5c4ea32e83402"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-04 16:32:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-16 00:48:41.000000000","tz":0},"subject":"Declare removal of Python 2 support","message":"Declare removal of Python 2 support\n\n... because Python 2 reached to its EOL while ago and is no longer\navailable since CentOS 8.\n\nChange-Id: Idb5fc37d9f334052839a0914895bf68f1b549037\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/10460e0c14bcbee6a455ce14131e4965d3e25c95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/10460e0c14bcbee6a455ce14131e4965d3e25c95"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[]}
