)]}'
{"id":"x%2Fpackstack~852194","triplet_id":"x%2Fpackstack~master~Ie7265335480848753e0576052db8754c4c25a517","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-16 10:38:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie7265335480848753e0576052db8754c4c25a517","subject":"setup.cfg: Replace dashes with underscores","status":"MERGED","created":"2022-08-04 16:34:06.000000000","updated":"2022-08-16 10:38:57.000000000","submitted":"2022-08-16 10:38:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852194","meta_rev_id":"73d08c8a89e5157139e4572712d5a48ec58306e7","_number":852194,"virtual_id_number":852194,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-16 10:38:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2022-08-16 09:54:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2022-08-16 09:54:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-04 17:14:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-16 09:54:34.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"51d04666ec29ca9ab99d300ad57a62ecd71b5455","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-04 16:34:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f35e99bd433048e9a18eb105fabcda0a746af1ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 17:14:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4dcfc67f7b864b38bbec3f9ea8f3fac0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5496c5ca7628433788ea397790ab861c : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f81efc0753b146548b58e17319f91601 : SUCCESS in 7m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff320f814a414e71be33d10dd7ea7df8 : SUCCESS in 7m 41s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/4fd4d7f046fd4c66951b2a2fd6f7f80b : FAILURE in 20m 59s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/3217636ae69d4f70914e2ac931cdf084 : FAILURE in 20m 20s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/31a7f497e9234a7c8f3c667798ad8356 : FAILURE in 19m 21s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/80bdb0583551430cba306b9e8a12aff4 : FAILURE in 24m 31s","accounts_in_message":[],"_revision_number":1},{"id":"bf2647b107767659b8ac13a99afe4124eb3efafb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-15 17:47:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8fe5ee06706707e69f6d8cf2f3b0b7e8cb0bcfd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 18:18:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f29db5015fa40aaa2a70f3bc9060747\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2ceba0b9ad724c0c96c53d3b307d4e4e : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff1f53be94634f7fad3cbe6bfe1a352d : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/816268a9671042ab9a6ee323a182fcb8 : SUCCESS in 4m 45s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/2365bf57cf724053bcbc5149b7902c75 : FAILURE in 24m 33s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/852a0cc8616e4cba8184029da031c8ff : FAILURE in 25m 21s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/28d5e698b844474dbe33ac0aac01f894 : FAILURE in 22m 33s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9f4e89a28a1649de92fd6b1a0ad3f744 : FAILURE in 22m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e05e8d8ab8b3f127356d3ef3cf5479e3c2a6cb44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-16 00:48:47.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ead72246a3893c8039ea2daafa909b0faa8cb312","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 01:51:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e5a1cc14de7416b9138dd646d01ec94\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/942e0842356a4e2ebd8e7252557adaf0 : SUCCESS in 1m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc42cf1f9a7a4f4d8d930d379507036e : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b626ff6552a447c87a5e43acb100194 : SUCCESS in 4m 59s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/ffb78e59e9bc408097eb17a45fd0ebae : SUCCESS in 30m 43s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/834e7dc716bc47469c171ad17dba7881 : SUCCESS in 54m 30s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/1054d1746095486b8b450ecae9ed0d57 : SUCCESS in 56m 09s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/60dd89b2e27e434792e83f69ebf6a8dc : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":2},{"id":"6c86de3c0627ce56d08ad11a5bad795403517de5","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-08-16 09:54:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"91b0aff9301f2c38191f9349840fd595992b0ed8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 09:55:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d1fbb02e505c968e9d59208f2a807c68dfc53b4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 10:38:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7e467ea1c7a428d8cd728502e2ad97a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a44132655595450dae4f56fc30a5ffa2 : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7159106f3574f2895a7239c5dd87599 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a53c48dfcd3849d990992354d88776d9 : SUCCESS in 3m 57s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/003153f6c8684b9eb058ab7a8fe5bf94 : SUCCESS in 41m 33s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/b4f03d1cfb724b4a8cc844fc24989118 : SUCCESS in 35m 19s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/32401a73e7504a598bd7c64460c1ecc2 : SUCCESS in 42m 49s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f9551cb249da48db8d81e77ec0820aef : SUCCESS in 39m 52s","accounts_in_message":[],"_revision_number":2},{"id":"73d08c8a89e5157139e4572712d5a48ec58306e7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 10:38:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3c454c4167a03f890adf471f9580dd24996553dd","revisions":{"a5031776d852e0cf86169e30b46a79464d4475ea":{"kind":"REWORK","_number":1,"created":"2022-08-04 16:34:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/852194/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/94/852194/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/94/852194/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/94/852194/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/94/852194/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/94/852194/1"}}},"commit":{"parents":[{"commit":"a8a16a38d3af73aeacb9b5d3d1531891e1610cfd","subject":"Declare removal of Python 2 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a8a16a38d3af73aeacb9b5d3d1531891e1610cfd"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-04 16:33:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-04 16:33:49.000000000","tz":540},"subject":"setup.cfg: Replace dashes with underscores","message":"setup.cfg: Replace dashes with underscores\n\nSetuptools v54.1.0 introduces a warning that the use of dash-separated\noptions in \u0027setup.cfg\u0027 will not be supported in a future version [1].\nGet ahead of the issue by replacing the dashes with underscores. Without\nthis, we see \u0027UserWarning\u0027 messages like the following on new enough\nversions of setuptools:\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\n[1] https://github.com/pypa/setuptools/commit/a2e9ae4cb\n\nChange-Id: Ie7265335480848753e0576052db8754c4c25a517\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a5031776d852e0cf86169e30b46a79464d4475ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a5031776d852e0cf86169e30b46a79464d4475ea"}]},"branch":"refs/heads/master"},"3c454c4167a03f890adf471f9580dd24996553dd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-08-16 00:48:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/852194/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/94/852194/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/94/852194/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/94/852194/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/94/852194/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/94/852194/2"}}},"commit":{"parents":[{"commit":"10460e0c14bcbee6a455ce14131e4965d3e25c95","subject":"Declare removal of Python 2 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/10460e0c14bcbee6a455ce14131e4965d3e25c95"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-04 16:33:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-16 00:48:47.000000000","tz":0},"subject":"setup.cfg: Replace dashes with underscores","message":"setup.cfg: Replace dashes with underscores\n\nSetuptools v54.1.0 introduces a warning that the use of dash-separated\noptions in \u0027setup.cfg\u0027 will not be supported in a future version [1].\nGet ahead of the issue by replacing the dashes with underscores. Without\nthis, we see \u0027UserWarning\u0027 messages like the following on new enough\nversions of setuptools:\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\n[1] https://github.com/pypa/setuptools/commit/a2e9ae4cb\n\nChange-Id: Ie7265335480848753e0576052db8754c4c25a517\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3c454c4167a03f890adf471f9580dd24996553dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3c454c4167a03f890adf471f9580dd24996553dd"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[]}
