)]}'
{"id":"x%2Fpackstack~853244","triplet_id":"x%2Fpackstack~stable%2Fyoga~Idb5fc37d9f334052839a0914895bf68f1b549037","project":"x/packstack","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-17 09:04:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idb5fc37d9f334052839a0914895bf68f1b549037","subject":"Declare removal of Python 2 support","status":"MERGED","created":"2022-08-16 08:57:02.000000000","updated":"2022-08-17 09:04:03.000000000","submitted":"2022-08-17 09:04:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853244","cherry_pick_of_change":852193,"cherry_pick_of_patch_set":2,"meta_rev_id":"30982f868edec791190c47fe52d8dc247e56a6a0","_number":853244,"virtual_id_number":853244,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-17 09:04:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2022-08-17 08:05:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2022-08-17 08:05:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-16 09:30:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-17 08:05:30.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"d6492834cd43820bf6ad324c28161ea41e6f07fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-16 08:57:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2d0c1186f53f2ef19809317b47d53d83f432a049","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 09:30:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/97e7c5d8506246c5b4d1abeed8c1ab42\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6b6a3be6c64f4153ac1791fe206e722a : SUCCESS in 2m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc0bc846db20439ba4000153ca63d43c : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb74803654564ea8acc93224aee0c18a : SUCCESS in 8m 31s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/4c14ef628cb2498db3b351fb5a5bd8bd : FAILURE in 21m 57s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/e0e3c65be7604bf09fe6aae6a7a96c8f : FAILURE in 15m 29s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/98d25ecd0efc4f7192f40e63ad6dcf3f : FAILURE in 14m 12s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e8e09f4a40734e4da8670f83a10d714f : FAILURE in 26m 23s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/7e4f6ce06e7f4f95b780c10fa7655833 : FAILURE in 19m 59s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/f59eedf3fd494de9b5067940b7c0baf3 : FAILURE in 14m 41s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/cf5c037b93794256afaf11da320c0c54 : FAILURE in 20m 16s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/b0f1c3e709794f17af823627a106a16c : FAILURE in 18m 46s","accounts_in_message":[],"_revision_number":1},{"id":"42a746ba9462f3b96e7402c3873c79c05727147f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-08-16 15:01:27.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"40ac03450453ff5069b389b270e89112b9de5ffa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 16:07:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cbae58dd717415786eb81612ea18491\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/81472925e1a4471a83c91268a26ba2b3 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2f9a39b69ef4e818e0f52c984c3d4ab : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/831ba403e0b7468f8ee4d483b732426a : SUCCESS in 4m 18s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/41a04219f163488d850da60edd16fc31 : SUCCESS in 52m 28s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/e1d396d621de4826926c34dc35fa028f : SUCCESS in 45m 45s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/39cfb2ef8480451ca492d1fe5ec1959e : SUCCESS in 1h 00m 29s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ab96e280c2f84369ab371767d43def14 : SUCCESS in 53m 42s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/98360e1aba8a40dd85f511f0f8c88bf0 : SUCCESS in 34m 26s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/f93a8f9bab45406ba99a1e638d124e33 : SUCCESS in 48m 07s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/53734bc9f97543fd8d524841ea5a5602 : SUCCESS in 50m 32s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/7342cc56f52d4caba0dc822274c68d7b : SUCCESS in 56m 02s","accounts_in_message":[],"_revision_number":2},{"id":"7c330cb0ac593accfea56a79aec449569cc3ca7a","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-08-17 08:05:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ef8eb443ed22bbf59b33fab0b817d65b7d6ac6d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 08:06:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"431dd491506dd63658cac0b304623e51db4f0dc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 09:04:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c56cf6567bf476aa2add19de0367be5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7f32c8dc350b41529b6a6754a7f3316a : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4340efe83e3b4f01be2bd5d821fdc662 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12c2a256a829446db1d6ca60239defc8 : SUCCESS in 4m 17s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/a633ab466ce4486fbe41905960577367 : SUCCESS in 41m 23s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/b6ddbd755a294713a84cf0fe9fd76414 : SUCCESS in 43m 13s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/e275d500d8724bc58ee32a93aa152fd8 : SUCCESS in 45m 27s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/2091d30bb2bb4816826043157e63820f : SUCCESS in 55m 48s","accounts_in_message":[],"_revision_number":2},{"id":"30982f868edec791190c47fe52d8dc247e56a6a0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 09:04:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0824a078cf705b807d0dcaaf3745ac518d3a5d87","revisions":{"b9bc171441d01cadd573dc57901e6bd65b6ce0c1":{"kind":"REWORK","_number":1,"created":"2022-08-16 08:57:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/853244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/44/853244/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/44/853244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/44/853244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/44/853244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/44/853244/1"}}},"commit":{"parents":[{"commit":"5105e00ba0646ea20c078f2502c9599e491f60ae","subject":"Mark Python 3.9 supported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5105e00ba0646ea20c078f2502c9599e491f60ae"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-04 16:32:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-16 08:57:02.000000000","tz":0},"subject":"Declare removal of Python 2 support","message":"Declare removal of Python 2 support\n\n... because Python 2 reached to its EOL while ago and is no longer\navailable since CentOS 8.\n\nChange-Id: Idb5fc37d9f334052839a0914895bf68f1b549037\n(cherry picked from commit 10460e0c14bcbee6a455ce14131e4965d3e25c95)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b9bc171441d01cadd573dc57901e6bd65b6ce0c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b9bc171441d01cadd573dc57901e6bd65b6ce0c1"}]},"branch":"refs/heads/stable/yoga"},"0824a078cf705b807d0dcaaf3745ac518d3a5d87":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-08-16 15:01:27.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/44/853244/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/44/853244/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/44/853244/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/44/853244/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/44/853244/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/44/853244/2"}}},"commit":{"parents":[{"commit":"2db39a442dec37b1eaa45d163467d8757ea520ba","subject":"Mark Python 3.9 supported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2db39a442dec37b1eaa45d163467d8757ea520ba"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-04 16:32:59.000000000","tz":540},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-08-16 15:01:27.000000000","tz":0},"subject":"Declare removal of Python 2 support","message":"Declare removal of Python 2 support\n\n... because Python 2 reached to its EOL while ago and is no longer\navailable since CentOS 8.\n\nChange-Id: Idb5fc37d9f334052839a0914895bf68f1b549037\n(cherry picked from commit 10460e0c14bcbee6a455ce14131e4965d3e25c95)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0824a078cf705b807d0dcaaf3745ac518d3a5d87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0824a078cf705b807d0dcaaf3745ac518d3a5d87"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[]}
