)]}'
{"id":"x%2Fpackstack~861683","triplet_id":"x%2Fpackstack~master~I9e6d2e21fcfb4847e4d54079cbb50e699931a07f","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-10-18 15:07:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9e6d2e21fcfb4847e4d54079cbb50e699931a07f","subject":"Ceilometer: Use built-in class to run ceilometer-upgrade","status":"MERGED","created":"2022-10-18 01:37:24.000000000","updated":"2022-10-18 15:07:35.000000000","submitted":"2022-10-18 15:07:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861683","meta_rev_id":"be4d7272b3f0d0a8f29434781aa7cbeb05833e07","_number":861683,"virtual_id_number":861683,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-18 15:07:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2022-10-18 13:34:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2022-10-18 13:57:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2022-10-18 13:34:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-18 02:41:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-18 13:34:37.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2022-10-18 13:57:39.000000000","updated_by":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"reviewer":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"state":"REVIEWER"}],"messages":[{"id":"2cd62191eefc366a1b845fa5c618aac3fb2025b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-18 01:37:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a51d17403c8184f31683f87e42207c90214f615","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 02:41:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f339fdee6d2427aa47a7f56054f3ab7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e05609436614724a155878a028795ee : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a97c3dfac2547059a47e40d02036aa3 : SUCCESS in 3m 08s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/554737e603c7430e99647f9f96e7850c : SUCCESS in 54m 32s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/76ad4d63e8d04ddea49bdb592e7a9b56 : SUCCESS in 52m 30s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/4d3b86f9207a49afa03b09fd6bf70f7d : FAILURE in 24m 12s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/20f651726e864a1c8f08cb034b5a4fdf : SUCCESS in 58m 35s","accounts_in_message":[],"_revision_number":1},{"id":"581a516050a9e000f65619d1546a8da77f4a4853","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-18 02:48:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7f5b830026700031bf22cf434d922348d634c59e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 03:47:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/385f364f2b534e198c4e62b29cdf2020\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/829b7a19399e4e10b72634aa86dbf375 : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f836363453341fcbd9886e9a5e61170 : SUCCESS in 3m 31s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/a826dadfe6e04a41b4ca278dfad848cf : SUCCESS in 48m 16s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/21bd08fdcd014ac1bdc402c4714afc2a : SUCCESS in 39m 24s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/32bfd0edea744ce08e9ad4ea72762081 : SUCCESS in 54m 23s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/071794b4bb614beeb2c068f6ffd5a4f3 : SUCCESS in 43m 12s","accounts_in_message":[],"_revision_number":2},{"id":"bfdf67bc2e7d6efe5b4929dec9ab21105dec268e","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-10-18 13:34:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"91950845da3b1f8ddee83ef9b8eb34143ad80d7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 13:35:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"78f6770c629ad74debe141428725a8e5f79ff2b1","author":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"date":"2022-10-18 13:57:39.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9bd757eac7e62de1c42aad4efd77b4dd59f31f25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 15:07:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ffc79383d924a17b7d5958ba05fa8f2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c634944ba5b644959fc5b82eeb507b30 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa6ad40edc0f41bfad480a2d75627ba4 : SUCCESS in 3m 24s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/97b584cc9558490cb6ac4c42c42d208f : SUCCESS in 37m 55s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/47efa5c43e884587bd766065c7df3360 : SUCCESS in 42m 50s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/b3bddc12c30a403980fdba1b9c1da274 : SUCCESS in 40m 55s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/669923c419804a079e2d5080f2e7b5ad : SUCCESS in 46m 29s","accounts_in_message":[],"_revision_number":2},{"id":"be4d7272b3f0d0a8f29434781aa7cbeb05833e07","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 15:07:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2c05591d9caed0f13acabbf56111a5bf1652c0f9","revisions":{"b01f4f2842d005e60c9e37599a8f59548c2d5d97":{"kind":"REWORK","_number":1,"created":"2022-10-18 01:37:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/861683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/83/861683/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/83/861683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/83/861683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/83/861683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/83/861683/1"}}},"commit":{"parents":[{"commit":"d044e10afc2e8f605a775c62d868cc767a497efd","subject":"Start redis before gnocchi-metricd when coordination uses redis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d044e10afc2e8f605a775c62d868cc767a497efd"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-18 01:36:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-18 01:36:58.000000000","tz":540},"subject":"Ceilometer: Use built-in class to run ceilometer-upgrade","message":"Ceilometer: Use built-in class to run ceilometer-upgrade\n\nThe puppet-ceilometer module provides the class to run the command, so\nyou can use the class instead of defining the raw exec resource.\n\nChange-Id: I9e6d2e21fcfb4847e4d54079cbb50e699931a07f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b01f4f2842d005e60c9e37599a8f59548c2d5d97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b01f4f2842d005e60c9e37599a8f59548c2d5d97"}]},"branch":"refs/heads/master"},"2c05591d9caed0f13acabbf56111a5bf1652c0f9":{"kind":"REWORK","_number":2,"created":"2022-10-18 02:48:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/861683/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/83/861683/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/83/861683/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/83/861683/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/83/861683/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/83/861683/2"}}},"commit":{"parents":[{"commit":"d044e10afc2e8f605a775c62d868cc767a497efd","subject":"Start redis before gnocchi-metricd when coordination uses redis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d044e10afc2e8f605a775c62d868cc767a497efd"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-18 01:36:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-18 02:47:55.000000000","tz":540},"subject":"Ceilometer: Use built-in class to run ceilometer-upgrade","message":"Ceilometer: Use built-in class to run ceilometer-upgrade\n\nThe puppet-ceilometer module provides the class to run the command, so\nyou can use the class instead of defining the raw exec resource.\n\nChange-Id: I9e6d2e21fcfb4847e4d54079cbb50e699931a07f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2c05591d9caed0f13acabbf56111a5bf1652c0f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2c05591d9caed0f13acabbf56111a5bf1652c0f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[]}
