)]}'
{"id":"x%2Fpackstack~861684","triplet_id":"x%2Fpackstack~master~I45b17579c2fcbd8f1f8027043b57ee3b14950831","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-10-20 13:09:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I45b17579c2fcbd8f1f8027043b57ee3b14950831","subject":"Ceilometer: Remove redundant installation of pythonN-redis","status":"MERGED","created":"2022-10-18 01:39:36.000000000","updated":"2022-10-20 13:09:16.000000000","submitted":"2022-10-20 13:09:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861684","meta_rev_id":"9031a2fab202f352555e4812206a52c67875d49c","_number":861684,"virtual_id_number":861684,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-20 13:09:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2022-10-20 12:06:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2022-10-20 12:06:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-18 02:26:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-20 12:06:07.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"513a5ea2d4cc82d83c036204560b0ac66fbf9d69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-18 01:39:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa77bfcd59b740a8846e033eacea9f12876ba8ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 02:26:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8307211c8d7541d3ad3aa081cb942274\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/509bad6ec4e547c7b78dfae933b44a2a : SUCCESS in 2m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a69c76ad02e147d2bb636beace94d934 : SUCCESS in 2m 54s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/94444b50621e4ce3b1c69e9a27c91e88 : SUCCESS in 40m 54s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d1fb9f6cba1c4fbb97924e23eb23be06 : SUCCESS in 28m 57s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ab57b60b968a4440865bff009ab4f75d : SUCCESS in 42m 55s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/1850f25274cb489db4cdae0a89af02ca : SUCCESS in 45m 03s","accounts_in_message":[],"_revision_number":1},{"id":"f7d29aae4bd6375fa950e23475b587f88a48a1fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-20 00:44:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7f71db27f61e9775835ff3468445fc77bb72546b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-20 00:46:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ee93010329d430dd38907c7c171ac1d810a9ec11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 01:52:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bc719a9a56a40d98171def70f47f42b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df93d2a934eb49ebb2e443eda20c2cba : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ea30af710cd417d98ddd3a0d39f9c03 : SUCCESS in 3m 17s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/1a80a353b1ce426eb4c12cb0b8d3bc2c : SUCCESS in 47m 19s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/ae320effc7a9444f89ad6c1486e5f1f0 : SUCCESS in 52m 11s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/746bda75d2c843fba5804542ee857183 : SUCCESS in 56m 37s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a8f350de7bf344c09612996cd91645f1 : SUCCESS in 1h 00m 46s","accounts_in_message":[],"_revision_number":3},{"id":"ad1af67f2a541eda7879296e2cdda7eabcca6119","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-10-20 12:06:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"befc063fce7a659ddcea04a71bd0d5a6444d0423","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 12:06:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0a349e29a52a11efad3f62c6e434343d316cf35a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 13:09:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0362a8e855d14ff6ad1f05468df4138c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d6cff62d32c4b299eecf926a2998e67 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff64eb449e82474caf1e82eb8eabc3da : SUCCESS in 3m 22s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/cc3044acc78c4b03bea74bb765bca105 : SUCCESS in 36m 12s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/6237aa4940af476489aa239a8a3b8367 : SUCCESS in 53m 10s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/c5d4b157ff804385a543ca3bb721eabe : SUCCESS in 57m 02s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/463beab15efd40d8978c444f1605f159 : SUCCESS in 42m 33s","accounts_in_message":[],"_revision_number":3},{"id":"9031a2fab202f352555e4812206a52c67875d49c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 13:09:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1f804e515bc12232730436d24eaae2025b28dc9c","revisions":{"984c114758dcf5c232323c3e58e76636d594a54f":{"kind":"REWORK","_number":1,"created":"2022-10-18 01:39:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/861684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/84/861684/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/84/861684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/84/861684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/84/861684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/84/861684/1"}}},"commit":{"parents":[{"commit":"d044e10afc2e8f605a775c62d868cc767a497efd","subject":"Start redis before gnocchi-metricd when coordination uses redis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d044e10afc2e8f605a775c62d868cc767a497efd"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-18 01:38:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-18 01:38:33.000000000","tz":540},"subject":"Ceilometer: Remove redundant installation of pythonN-redis","message":"Ceilometer: Remove redundant installation of pythonN-redis\n\nThe pythonN-redis is indirectly required by the ceilometer-common\npackage (ceilmeter-common -\u003e pythonN-tooz -\u003e pythonN-redis) so we no\nlonger need to install the package explicitly.\n\nChange-Id: I45b17579c2fcbd8f1f8027043b57ee3b14950831\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/984c114758dcf5c232323c3e58e76636d594a54f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/984c114758dcf5c232323c3e58e76636d594a54f"}]},"branch":"refs/heads/master"},"3ef309607dbc1a01e7fff7c469d76f7ea1766de8":{"kind":"REWORK","_number":2,"created":"2022-10-20 00:44:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/861684/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/84/861684/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/84/861684/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/84/861684/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/84/861684/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/84/861684/2"}}},"commit":{"parents":[{"commit":"535aaf736d9d72a3e142e36e0e11d0458a157fe5","subject":"Merge \"Establish order between redis and aodh and ceilometer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/535aaf736d9d72a3e142e36e0e11d0458a157fe5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-18 01:38:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-20 00:44:03.000000000","tz":540},"subject":"Ceilometer: Remove redundant installation of pythonN-redis","message":"Ceilometer: Remove redundant installation of pythonN-redis\n\nThe pythonN-redis is now indirectly required by the ceilometer-common\npackage (ceilmeter-common -\u003e pythonN-tooz -\u003e pythonN-redis) so we no\nlonger need to install the package explicitly.\n\nChange-Id: I45b17579c2fcbd8f1f8027043b57ee3b14950831\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3ef309607dbc1a01e7fff7c469d76f7ea1766de8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/3ef309607dbc1a01e7fff7c469d76f7ea1766de8"}]},"branch":"refs/heads/master"},"1f804e515bc12232730436d24eaae2025b28dc9c":{"kind":"REWORK","_number":3,"created":"2022-10-20 00:46:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/861684/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/84/861684/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/84/861684/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/84/861684/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/84/861684/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/84/861684/3"}}},"commit":{"parents":[{"commit":"535aaf736d9d72a3e142e36e0e11d0458a157fe5","subject":"Merge \"Establish order between redis and aodh and ceilometer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/535aaf736d9d72a3e142e36e0e11d0458a157fe5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-18 01:38:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-20 00:46:29.000000000","tz":540},"subject":"Ceilometer: Remove redundant installation of pythonN-redis","message":"Ceilometer: Remove redundant installation of pythonN-redis\n\nThe pythonN-redis is now indirectly required by the ceilometer-common\npackage (ceilmeter-common -\u003e pythonN-tooz -\u003e pythonN-redis) so we no\nlonger need to install the package explicitly.\n\nChange-Id: I45b17579c2fcbd8f1f8027043b57ee3b14950831\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1f804e515bc12232730436d24eaae2025b28dc9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1f804e515bc12232730436d24eaae2025b28dc9c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[]}
