)]}'
{"id":"x%2Fpackstack~861685","triplet_id":"x%2Fpackstack~master~I010570cbf5a56629c532b726bf4c2db20615b800","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-10-18 16:51:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I010570cbf5a56629c532b726bf4c2db20615b800","subject":"Drop firewall rule for ceilometer-api","status":"MERGED","created":"2022-10-18 01:43:38.000000000","updated":"2022-10-18 16:51:07.000000000","submitted":"2022-10-18 16:51:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861685","meta_rev_id":"0ff833868b327d5cb158b03df30579e47e8b5c50","_number":861685,"virtual_id_number":861685,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-18 16:51:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2022-10-18 13:32:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2022-10-18 13:57:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2022-10-18 13:32:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-18 02:48:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-18 13:32:42.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2022-10-18 13:57:53.000000000","updated_by":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"reviewer":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"state":"REVIEWER"}],"messages":[{"id":"7e8ec54a633638e7e1815dd4a2db920ff3bb77ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-18 01:43:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aab81257f64376535018aa04dbdfc324aa3d6ff1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-18 01:44:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9fa8ad175bd41b6011c96fc5f919781599cf499a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 02:48:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8535db45bde5471581011e723ce62d12\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce50dab67789433b88dd4200c438df3b : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2703a70a11344de8933658cd164beb7 : SUCCESS in 3m 14s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/199c9fb117fe4de997ecc85d2820cdee : SUCCESS in 49m 51s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/310e7e594f11406ca8f12d9812a15af4 : SUCCESS in 52m 12s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/1e967b3c77b645f790b7c4cba59d1d72 : SUCCESS in 55m 49s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ca371fdcfedf4631b662b8803a84ed79 : SUCCESS in 57m 39s","accounts_in_message":[],"_revision_number":2},{"id":"ea1c2603de788b2057986f5fcfe886af27ef3526","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-10-18 13:32:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7b9ef2f0590c8b041ab537ace4359f39ac8dbedb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 13:33:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f3204a689d328010be3da3af6db0a703717a1923","author":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"date":"2022-10-18 13:57:53.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d54a671ba9890609d85693b01a9bb3a6811f5563","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 14:19:34.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae1c32572d174a89862a0af6370aa13a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c45823f472b948bb955ed138180d1261 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5de99458de544d50a90dfc9fb25580b3 : SUCCESS in 4m 19s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/3934a24b5f8440a7bef3ea8634e931c7 : SUCCESS in 39m 57s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/93ca7a24d93e45e7abc74626fbd9e835 : FAILURE in 39m 43s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/0936749457224b6991a232d404926fca : SUCCESS in 43m 17s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e553d5733a4146f39ffdbefab6ddcade : SUCCESS in 45m 09s","accounts_in_message":[],"_revision_number":2},{"id":"ec925ffb41206ff78be7fd9e165bec1556af5fb1","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-10-18 14:57:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dcce4058ae4922e6a1d79260dec4a3b2dc3a9f85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 16:06:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6ae8339e62a441a9aa42d756c6c8f5a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3899e3c9660f45dda965669604acba13 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fea4c557984476a98f3d6584290fe2d : SUCCESS in 3m 31s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/c30d58f7d9534997b372d9b039dde237 : SUCCESS in 47m 22s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/16d10471553641c3916db7090953fbb9 : SUCCESS in 48m 17s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/036a2793aace45baa1eed8b6c81fc4d5 : SUCCESS in 1h 00m 02s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/0abd760e9cb340638dbac9651af9cb05 : SUCCESS in 1h 03m 12s","accounts_in_message":[],"_revision_number":2},{"id":"69daf6d27ad7117d65c97e7f23d67d11a2feb7fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 16:07:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c012ab752aac2b6ea822de3f824040ec0c6ea05f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 16:51:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07e3696a64b64ccbaa0c4377e245dae8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bed78a89678494aabb8e0c3dd428cb9 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/719c481f3dc24901bcf39b13796289d1 : SUCCESS in 3m 10s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/ad195351cc78449f91e951c4440e1ccb : SUCCESS in 36m 22s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/dcc4a3e1e78842f097b1380ab5dded9c : SUCCESS in 38m 54s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/57cbab28bc1942c0bd86b917d7db8547 : SUCCESS in 39m 11s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d680d3f8050943ca935cd3583ba357d6 : SUCCESS in 43m 02s","accounts_in_message":[],"_revision_number":2},{"id":"0ff833868b327d5cb158b03df30579e47e8b5c50","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 16:51:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bcad045b42d1a6e5f205fef66795174515dc23f2","revisions":{"0a8411ab4890ac760092b06d691ec545540e1659":{"kind":"REWORK","_number":1,"created":"2022-10-18 01:43:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/861685/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/85/861685/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/85/861685/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/85/861685/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/85/861685/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/85/861685/1"}}},"commit":{"parents":[{"commit":"d044e10afc2e8f605a775c62d868cc767a497efd","subject":"Start redis before gnocchi-metricd when coordination uses redis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d044e10afc2e8f605a775c62d868cc767a497efd"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-18 01:42:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-18 01:42:27.000000000","tz":540},"subject":"Drop firewall rule for ceilometer-api","message":"Drop firewall rule for ceilometer-api\n\n... because the ceilometer-api service was removed long ago.\n\nChange-Id: I010570cbf5a56629c532b726bf4c2db20615b800\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0a8411ab4890ac760092b06d691ec545540e1659"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0a8411ab4890ac760092b06d691ec545540e1659"}]},"branch":"refs/heads/master"},"bcad045b42d1a6e5f205fef66795174515dc23f2":{"kind":"REWORK","_number":2,"created":"2022-10-18 01:44:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/861685/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/85/861685/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/85/861685/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/85/861685/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/85/861685/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/85/861685/2"}}},"commit":{"parents":[{"commit":"d044e10afc2e8f605a775c62d868cc767a497efd","subject":"Start redis before gnocchi-metricd when coordination uses redis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d044e10afc2e8f605a775c62d868cc767a497efd"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-18 01:42:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-18 01:44:04.000000000","tz":540},"subject":"Drop firewall rule for ceilometer-api","message":"Drop firewall rule for ceilometer-api\n\n... because the ceilometer-api service was removed long ago.\n\nChange-Id: I010570cbf5a56629c532b726bf4c2db20615b800\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bcad045b42d1a6e5f205fef66795174515dc23f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bcad045b42d1a6e5f205fef66795174515dc23f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[]}
