)]}'
{"id":"x%2Fpackstack~865810","triplet_id":"x%2Fpackstack~master~Iffb87bfda58e38809e244ea84eb925fe3aec3c25","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-11-29 12:22:53.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2022-11-29 12:40:26.000000000","reason":"\u003cGERRIT_ACCOUNT_16312\u003e replied on the change","reason_account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}},"hashtags":[],"change_id":"Iffb87bfda58e38809e244ea84eb925fe3aec3c25","subject":"Remove enable_ssl from keystone.pp","status":"MERGED","created":"2022-11-28 09:36:30.000000000","updated":"2022-11-29 13:29:44.000000000","submitted":"2022-11-29 13:29:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865810","meta_rev_id":"cbea4eae25bf6c62bc11c5592c7c8d30038b9afa","_number":865810,"virtual_id_number":865810,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-29 13:29:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2022-11-29 12:22:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2022-11-28 09:37:01.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2022-11-29 10:19:00.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2022-11-29 12:40:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-28 09:37:01.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-11-28 10:39:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-29 10:19:00.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"},{"updated":"2022-11-29 10:24:33.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"7630b6cb564d9df46a92f7b3146d44d84f3f3973","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-11-28 09:36:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8b1bde64a9af37c21c8997eef43f84cdfc284c0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-28 09:37:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a68c3f5add09aaddc7c7b644cba3cf51b077637b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 10:39:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba166b06c97d444e802381b71074fb47\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1821644e3448455baf081e7559e2cb7b : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5df007514924a5baf5d9d47ef44b021 : SUCCESS in 3m 43s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/8ca1088a4ac84d4f8f53a89c9d9a0ce7 : FAILURE in 53m 25s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/edacb3b5da8a4044b9a3e54e35b7c216 : SUCCESS in 56m 51s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/0fbf550805334811a5703e543acdd9b8 : SUCCESS in 57m 24s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9c2b591f4c76439db71f192442657fdc : SUCCESS in 45m 19s","accounts_in_message":[],"_revision_number":1},{"id":"6d2c5f674e96ac84f21d99b4add6738abe28be1a","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-11-28 11:19:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"87d3570290c2497a636c17ed73df7aaed3a8fa46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 12:05:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5b99774b717422a9eed9a6df472571a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b97a4dde5e644680a9f58c342168026e : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b102a1bf96d0402f8a339c4b624b13db : SUCCESS in 4m 14s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/9e4e68f7ef894274b84e242b1ff1e093 : FAILURE in 41m 01s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/c93e2446c9534e98a371bbfc4102e1c8 : SUCCESS in 40m 18s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/8c1fb2ed855e42d9a31d096499084195 : SUCCESS in 41m 55s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/cfc92c0915f24b01b33a8170d56a065f : SUCCESS in 40m 46s","accounts_in_message":[],"_revision_number":1},{"id":"3c7aff259437e13c8cffe2b19325769a43756d79","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-28 15:06:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3de0b3b4011e5a5fdd966c347e61fac81f00b927","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 16:19:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/600eaf00a3e74576a17167315cd7a021\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57bbf70bff664578bb4a7282f012ad16 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8257fec1183245749d3c9b7719a3dfdf : SUCCESS in 3m 15s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/5b16a8475e1845f196fac316e7adaf7c : SUCCESS in 54m 56s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/24343420ba0642fda8e8bf1630bbf708 : SUCCESS in 48m 38s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/be77a8c2f5254dc58a25349aeee02d35 : SUCCESS in 50m 19s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/352e9d49efa04616b327dfba0707636e : SUCCESS in 1h 04m 44s","accounts_in_message":[],"_revision_number":1},{"id":"98f05b843cc668c803eef35dd7db28786a44748a","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2022-11-29 10:19:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"32705f919006ddd8a19177ee3849b71f6c52adfc","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-11-29 12:22:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"622f9a05cd9f3920e32084cba83dd44a6177f1bc","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-11-29 12:40:26.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dea07f6b6aab5af90981859034f60cedb3a019cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 12:41:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"17bafd24483e408cec7de88e8d5a5c48129c4ee3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 13:29:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a6d2191fb324acbbf9a62d8f45ec290\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc142ed604d6417ab5e4cc16c34b28d3 : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1e2e1fb89ba4fdea6e30af4a737fb9d : SUCCESS in 2m 53s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/991119b442b74f08888c79805e63801a : SUCCESS in 33m 51s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/3f399d7cec4849a59f10894bf11c5a4c : SUCCESS in 35m 54s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/5651dbf622d34e85ac20219090e3be27 : SUCCESS in 35m 59s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/cf7d2faab9ff46b68dc8437991368b2f : SUCCESS in 41m 59s","accounts_in_message":[],"_revision_number":1},{"id":"cbea4eae25bf6c62bc11c5592c7c8d30038b9afa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 13:29:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e8078a4f4077f5492236654e00e2729d19a3f23d","revisions":{"e8078a4f4077f5492236654e00e2729d19a3f23d":{"kind":"REWORK","_number":1,"created":"2022-11-28 09:36:30.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/10/865810/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/10/865810/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/10/865810/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/10/865810/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/10/865810/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/10/865810/1"}}},"commit":{"parents":[{"commit":"1f804e515bc12232730436d24eaae2025b28dc9c","subject":"Ceilometer: Remove redundant installation of pythonN-redis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1f804e515bc12232730436d24eaae2025b28dc9c"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-11-28 09:34:44.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-11-28 09:36:28.000000000","tz":60},"subject":"Remove enable_ssl from keystone.pp","message":"Remove enable_ssl from keystone.pp\n\nIt has been removed in puppet-keystone [1] and actually it has no effect\nsince some releases ago as the ssl termination is managed in the wsgi\nserver.\n\n[1] https://review.opendev.org/c/openstack/puppet-keystone/+/864675\n\nChange-Id: Iffb87bfda58e38809e244ea84eb925fe3aec3c25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e8078a4f4077f5492236654e00e2729d19a3f23d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e8078a4f4077f5492236654e00e2729d19a3f23d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[]}
