)]}'
{"id":"x%2Fpackstack~865971","triplet_id":"x%2Fpackstack~stable%2Fzed~Idc8d8ce8fe2d8d5dddc1f7dad0546e880759a094","project":"x/packstack","branch":"stable/zed","topic":"zed","attention_set":{},"removed_from_attention_set":{"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2023-03-02 11:13:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idc8d8ce8fe2d8d5dddc1f7dad0546e880759a094","subject":"Adapt for stable/zed","status":"MERGED","created":"2022-11-29 10:44:51.000000000","updated":"2023-03-02 11:13:27.000000000","submitted":"2023-03-02 11:13:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865971-zed","meta_rev_id":"b75c7a68afd00f39e4bbca93091609d15ccdf429","_number":865971,"virtual_id_number":865971,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-02 11:13:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2023-03-02 06:14:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2023-03-02 07:20:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-29 11:41:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-02 06:14:21.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"fcbb7e0b5c1cf3784f79c8fdacd6f8ddd9099fa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-11-29 10:44:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"78bbb4fe4e979dee42efe12d504aee79b8dc5b29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 11:41:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c54ee14e5664fb1ae7f7640b4b8152a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a36a813163d45f9bf65fe77a966acc8 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2cd53e2166c4213963bcf12db0e4233 : SUCCESS in 4m 16s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/05354d92433c47d6855713c9501e1b20 : FAILURE in 38m 08s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/7186a1dfff1844a7b16f24a2d2e73f38 : SUCCESS in 40m 09s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/2f02dd21bfec411faf1be402c41d2fdd : SUCCESS in 36m 59s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f02fd87ddc844a7fb8b1dd2803941f19 : SUCCESS in 52m 10s","accounts_in_message":[],"_revision_number":1},{"id":"a488ef61f4ca8c4e344b0647578f582e1cdfd57e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-03-02 06:14:21.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e39a3c82434af206f0725df15854b44f3b5b72a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 07:14:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48fa769b9293492686bc51aaa0bcd642\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52d87e6c05fa437294f6e6a3b271e4af : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7a0ea209e5d4e41b40ff2c9a4114cb2 : SUCCESS in 2m 48s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/00d101a6c28f4d858c995cf83ac056e2 : SUCCESS in 52m 57s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/57e9d86fec8b4e2191184f4c9164beb7 : SUCCESS in 46m 44s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a21d758d809c4933a28ce0ce3f7f4e9e : SUCCESS in 50m 39s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e837a593761f4659957f1fd885b6c50d : SUCCESS in 53m 41s","accounts_in_message":[],"_revision_number":1},{"id":"ecf3564f41cadb46388ae951803e9ec2693dbfc4","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-03-02 07:20:15.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"da084a7561c3bd8abd9c9ae2900582989c716272","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 07:21:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb164728691008a99917e9fb6ca640c9403e7351","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 08:32:01.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a907386c68a248cf846fd852d516c057\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56c3f98f201f4475bbebaed3d3f57b5e : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6550132226194f7799c8e50511d4f15b : SUCCESS in 3m 28s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/bfcfe5aa470044418bf41c12fbad0de3 : FAILURE in 23m 14s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/f78734eabf724772be6457e8a6a4622c : FAILURE in 53m 03s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/70d0db0a75264ad483be8920bd5844ef : SUCCESS in 58m 16s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/58bfc8c7531743b19c0b4a29692af70f : SUCCESS in 1h 04m 06s","accounts_in_message":[],"_revision_number":1},{"id":"d662a8f3080846276b84225513bbbb64a81aa6e2","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-03-02 08:34:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29e9318d5cdf30113cefb7bfe5c4fb39c890c843","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 10:05:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0512f9337ca542258a0a770251b2c9f6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51bef7816ef14ab6b821fe640f5e4d3e : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c30a79ce6afb4f00bac7f96cfae7d4bf : SUCCESS in 3m 36s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/7ed6a8f2cc534d858a941a1e8d6a02f4 : SUCCESS in 51m 53s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/83b736e60ec349628d992b707fde8d2d : SUCCESS in 54m 56s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/300316127e724a65b18a1891a6f098dc : SUCCESS in 46m 45s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/55276db770344b119b45273517bcbcf2 : SUCCESS in 51m 21s","accounts_in_message":[],"_revision_number":1},{"id":"1ad723ccccea06b7c85a7807b862c2abd7ffd813","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 10:06:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed6f0f69272c149f1c4b129aae4de2c80eb950df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 11:13:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ebd734b3efa468c849acda2ca7bbfe4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb982911006c4511940121736273ed7b : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccba809d554a490a9e53758dedafab89 : SUCCESS in 3m 21s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/a5d00d04311e4950b756e1f015d93dd5 : SUCCESS in 55m 55s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/3153519268b94b2fb08518fb6ed85805 : SUCCESS in 33m 59s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/e9d81b6b06414820ba5b6e789afd235b : SUCCESS in 52m 49s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/65de9e1bb85b4faabe2691342a76e5ff : SUCCESS in 1h 02m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b75c7a68afd00f39e4bbca93091609d15ccdf429","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 11:13:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"012f9e4fbf86673e1642c225ef85329a735bbda0","revisions":{"012f9e4fbf86673e1642c225ef85329a735bbda0":{"kind":"REWORK","_number":1,"created":"2022-11-29 10:44:51.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/71/865971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/71/865971/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/71/865971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/71/865971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/71/865971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/71/865971/1"}}},"commit":{"parents":[{"commit":"1f804e515bc12232730436d24eaae2025b28dc9c","subject":"Ceilometer: Remove redundant installation of pythonN-redis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1f804e515bc12232730436d24eaae2025b28dc9c"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-11-29 10:43:36.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-11-29 10:44:50.000000000","tz":60},"subject":"Adapt for stable/zed","message":"Adapt for stable/zed\n\n- Fix .gitreview\n- Using puppet openstack modules from stable/zed\n- Using non-openstack puppet modules from pinned commit hashes in RDO\n- Using rpm repos from zed instead of master\n\nChange-Id: Idc8d8ce8fe2d8d5dddc1f7dad0546e880759a094\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/012f9e4fbf86673e1642c225ef85329a735bbda0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/012f9e4fbf86673e1642c225ef85329a735bbda0"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}}]}],"submit_requirements":[]}
