)]}'
{"id":"x%2Fpackstack~866548","triplet_id":"x%2Fpackstack~master~I8fc4504a76e9bb0725a60f3bcfab8355b4f52ac4","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-04-11 12:03:39.000000000","reason":"Change was submitted"},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2023-04-11 07:14:51.000000000","reason":"\u003cGERRIT_ACCOUNT_16312\u003e replied on the change","reason_account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}},"hashtags":[],"change_id":"I8fc4504a76e9bb0725a60f3bcfab8355b4f52ac4","subject":"Heat: Use separate class to manage [trustee] options","status":"MERGED","created":"2022-12-05 07:15:17.000000000","updated":"2023-04-11 12:03:39.000000000","submitted":"2023-04-11 12:03:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866548","meta_rev_id":"7ff6a5da11c49c88b66611d92e54cdb126816fd7","_number":866548,"virtual_id_number":866548,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-11 12:03:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2023-04-11 07:14:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2023-04-11 07:14:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-05 07:53:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-15 03:16:28.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"93656314cc9c8b33d58cbdbfcc208089a9c942c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-05 07:15:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4e8118d6716ec7d2f4a009d2e882b4048293beb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-05 07:53:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0997d31cf95493bb61c6fa6ab1d142e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/100e72b4314340dfa0c7088d4cbbac79 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0493ea25265447db8057097b60d0ad0 : SUCCESS in 3m 27s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/1c46c6815dd041d9a1b62235780449f3 : SUCCESS in 31m 34s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/6dbb7a9087df4646bb58a9ad6ee8737c : FAILURE in 33m 50s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/c134909607f34bcdbfd4100566f54305 : SUCCESS in 35m 20s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/09c7b94898f24b9e86f1066acb47ef6c : SUCCESS in 35m 55s","accounts_in_message":[],"_revision_number":1},{"id":"2232fd79a27d468a0f2014c6bf29723389bc61f1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-05 08:04:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9b77909a63e94f609e91c5f0060934961a06f460","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-05 09:10:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/004dbaa818154260827cafd387b83cfa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/907be8624afc42f195d914f3a1c9f76f : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce4d7445296c4f71aa76a0b53ea9ceb7 : SUCCESS in 3m 07s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/90ae0107db8e4125ac9e6d375c32045b : SUCCESS in 52m 46s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/ff93fc281df04ccc9adbdc7d12deb310 : SUCCESS in 52m 07s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/510948626f394738b27028c506d5656c : SUCCESS in 56m 54s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ac4a2caba7f2484ebc4a9b120b4003ea : SUCCESS in 1h 00m 37s","accounts_in_message":[],"_revision_number":1},{"id":"c729b480b8eca57b837bb2c3db6d6c8d73c011d9","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2023-04-11 07:14:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e183a540949148f2da1e6e694c0369f50af6c76c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 07:15:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a85757db9fba0b1c24c761263a27636aa825d795","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 08:11:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/53cbed06bd30464dab3f57a9235883fc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3c45ae7258246d39f22400c0663bfe9 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4b3b7943de8461aacdc7aae072d2b5e : SUCCESS in 5m 57s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/ced05d9a670740a689fbb18dc3eac907 : FAILURE in 29m 10s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/34137f6d75f74a55a1b2b898a8a088fa : SUCCESS in 43m 02s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a3ae3c46e7464ea997140b8777a4bcb0 : SUCCESS in 44m 35s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/4cdda3ced2f643189e0bd0c88da67a41 : SUCCESS in 46m 27s","accounts_in_message":[],"_revision_number":1},{"id":"be30e9b4d03028fa7060b913ca582244c3a2a5ff","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2023-04-11 08:18:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"810cf22630da3883d0fb7eb6d75d1fb1be9d503a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 09:17:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b136cc9617e24ebcb9296e285881fd94\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d025bf6b222e456e877f63eafb83a9e3 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f64d866c82544be690e43f91b20e0031 : SUCCESS in 4m 00s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/b3182f4e16f24a77b27415c13361b0c1 : SUCCESS in 38m 08s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/2e30909ec0f54d5597fa32235ed50a4f : SUCCESS in 38m 27s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/7c3f2f6cff0c4e8a9c7a03d0284abc80 : SUCCESS in 42m 23s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/23eddd35617c423686463d38e8f364e1 : SUCCESS in 44m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b57acffa05db1bb0dd8462d216f469cbd09d532e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 09:18:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"07d69d7a0e0827ed6b0cbe54a4c4218737b809dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 10:00:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4dbb57a8da5343c69d0adfb08e6e9ed6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f7804505755489fa678fb4d54cd502f : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5c3f32c77fb49cab8bee78525354e87 : SUCCESS in 4m 04s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/14256a321cff4221be26ba2f6361c164 : SUCCESS in 34m 30s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/ba149dd32d6e40f88a8fcc8ea87298bd : FAILURE in 38m 01s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/9f0386b0f0e44f4f885b8721b58daca1 : SUCCESS in 39m 01s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f31afeaf1a8d4dc5809b30a226e93528 : SUCCESS in 41m 25s","accounts_in_message":[],"_revision_number":1},{"id":"be6b95e2bf93e2f2870f848b12b5c2770563728e","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2023-04-11 10:13:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc22631bcb3829de8bb30607bac78e99f456dfe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 10:57:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/516ab1b1ca8741feaa99c38edbb598fe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e6706f9acb046df8ef0e60c19a493da : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50a4f36d10fe44f396dbb85b3fc49e23 : SUCCESS in 3m 28s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/707d809ad2d047639e3685a4f9558d56 : SUCCESS in 36m 58s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/b419720ae1b845d2896426e3a060adca : SUCCESS in 37m 20s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/545a942898a542ed8e9d371dc0bfde08 : SUCCESS in 36m 54s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/521d8d4d361d4a4b87c6865a1f312efa : SUCCESS in 39m 24s","accounts_in_message":[],"_revision_number":1},{"id":"93180d3238c065ed63282f13171d631c869169e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 10:58:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"44ac7f230534094b0eb942a1199362cb2cf3c51e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 12:03:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d287a6f6efd4f21aa28980a821cbcc0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19c18a98280b47a8be606fb9d1f3bf7b : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1dd1e48848743508bcdaf84887d0969 : SUCCESS in 3m 28s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/41b19f7959244be39a35e2174a4b3483 : SUCCESS in 34m 28s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/2c338b4384a543e393eac6a4d5f1ba0a : SUCCESS in 35m 15s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/7717766fe25e408e83ae4bce70110663 : SUCCESS in 41m 24s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/03ea03b2cffa45a09ab8db4da951641c : SUCCESS in 40m 58s","accounts_in_message":[],"_revision_number":1},{"id":"7ff6a5da11c49c88b66611d92e54cdb126816fd7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 12:03:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"261bc218e7201382fcdd0e51b31f1e33ea87c9d5","revisions":{"261bc218e7201382fcdd0e51b31f1e33ea87c9d5":{"kind":"REWORK","_number":1,"created":"2022-12-05 07:15:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/866548/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/48/866548/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/48/866548/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/48/866548/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/48/866548/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/48/866548/1"}}},"commit":{"parents":[{"commit":"1f804e515bc12232730436d24eaae2025b28dc9c","subject":"Ceilometer: Remove redundant installation of pythonN-redis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1f804e515bc12232730436d24eaae2025b28dc9c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-05 07:13:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-05 07:14:55.000000000","tz":540},"subject":"Heat: Use separate class to manage [trustee] options","message":"Heat: Use separate class to manage [trustee] options\n\nThis change ensures the heat::trustee class, which was introduced\nduring Yoga cycle[1], is used to manage the [trustee] options.\n\n[1] 6e8b799ba8f8761616d52928e479a914ef899ea2\n\nChange-Id: I8fc4504a76e9bb0725a60f3bcfab8355b4f52ac4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/261bc218e7201382fcdd0e51b31f1e33ea87c9d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/261bc218e7201382fcdd0e51b31f1e33ea87c9d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
