)]}'
{"id":"x%2Fpackstack~875494","triplet_id":"x%2Fpackstack~master~Ie70d0f242220b513bab37befedb9ab785e08d077","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-03-01 08:15:55.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2023-03-01 15:22:32.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-02-28 01:17:31.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2023-03-01 08:17:45.000000000","reason":"\u003cGERRIT_ACCOUNT_31068\u003e replied on the change","reason_account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}},"hashtags":[],"change_id":"Ie70d0f242220b513bab37befedb9ab785e08d077","subject":"Enable snapshots in NFS backends and add nova section","status":"MERGED","created":"2023-02-27 13:13:47.000000000","updated":"2023-03-01 15:22:32.000000000","submitted":"2023-03-01 15:22:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875494","meta_rev_id":"93c53cffdc7b24fbc7b67d57a2972af373cdbcb6","_number":875494,"virtual_id_number":875494,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-01 15:22:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2023-03-01 08:15:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2023-02-28 01:17:31.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2023-03-01 08:17:45.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2023-03-01 11:53:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-27 15:42:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-27 17:39:11.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-02-27 17:39:11.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2023-02-27 17:39:11.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"}],"messages":[{"id":"bddcef283b1d367538227754f3841e5d5b541b43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2023-02-27 13:13:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b870b3d3a76b64f1249314d8529b07b5c5c8cbb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2023-02-27 14:37:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a97928d05bb94b142d633c05871319e0fbe993bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 15:42:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c10cec5c31c447bb26f5e8b36df406b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c95a8d5fa2104849aa884dd5fa57453c : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3ba45090e904d7f8cad166036b9d066 : SUCCESS in 4m 00s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/a5bbeea66a0645c9843c43e44b0a41bc : SUCCESS in 39m 16s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/e960f1aefcc44050928311468f37acba : SUCCESS in 43m 40s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/df04c8ec335b4cdb97a30311ebc81d38 : SUCCESS in 40m 27s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a100a3604f6443fd9fa25456931c00e7 : SUCCESS in 43m 09s","accounts_in_message":[],"_revision_number":2},{"id":"398403d0164e2199256a9ace7fc2e3235b815406","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-28 01:17:31.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3023c0e23d4d0a294dc36b0bfe797dccfafc7e4a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-03-01 08:15:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"62d7209d12a0f92b8b2b8a2dfb4bd37d68095064","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2023-03-01 08:17:45.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"136d8100c83c33761fef82b3e323c62ab7dbc141","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-03-01 11:53:37.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"49d8d9ce6c12681dd2b84c53f7ae794a78df5e3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 11:54:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b4569476214d226a5bbab16e86de65821a510f5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 12:59:42.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f062b8efe8f4ff2949b868c3b106d4a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/818593b1638e42c59e8bd1e7c79f8e99 : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1934714c99684a35b9011c87642128e2 : SUCCESS in 3m 07s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/8f6046adf34a493780e980fd34596364 : SUCCESS in 48m 42s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/aab815de2c0b490f8f4bd9676c952ec3 : SUCCESS in 52m 47s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/f5dabc4167b8468f9199b33f8e31aff1 : SUCCESS in 51m 05s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/635861663398406aadd6f32b8eab7e36 : FAILURE in 59m 29s","accounts_in_message":[],"_revision_number":2},{"id":"118d5b56fd370453eb3f4b14b92b24a9ffda52cc","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-03-01 13:02:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d6d9bc9b7a2faf4c684fb583465ec2a8d93d7d7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 14:26:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/712cf6f081e84b66a1cd8805c4cde939\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6a13ea56d304172b425a8c064e55d60 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51b5ac7bf72e45868c5303033810dd03 : SUCCESS in 4m 19s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/3c4ae138fc9c48dc8578ea33047727c9 : SUCCESS in 46m 06s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/0566e29f005949b2b3fa1f2a478b73ef : SUCCESS in 39m 49s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/d95fb66a2901409bb4b190166f78e123 : SUCCESS in 50m 46s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/917d5e58f51e4cfe83d182d8aafdeaeb : SUCCESS in 1h 17m 05s","accounts_in_message":[],"_revision_number":2},{"id":"c76f489d95f7154fd0d63fd842aa39f6dcbe0a00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 14:27:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f5ce29ff17a86c08df35d644447b1a8158036545","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 15:22:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/580b8b35ef584db39803bff00c6906f9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f9c1055220d466182c34a7928f058ff : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f4f2ae3b49f42c4948b1867fb1cf690 : SUCCESS in 2m 58s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/bd0d4e22407942999480900f5a015fdb : SUCCESS in 39m 25s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/e64da08c2f1541779724f60296e502d9 : SUCCESS in 41m 04s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/2119092a1c2640c6aae282dc4d0bd9b2 : SUCCESS in 45m 32s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/1fd2b3021c8347da8aa11836462c15a2 : SUCCESS in 50m 51s","accounts_in_message":[],"_revision_number":2},{"id":"93c53cffdc7b24fbc7b67d57a2972af373cdbcb6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 15:22:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bd1e892301bfbb392c870a314dcbbb404d5a1e99","revisions":{"ee2b4b668786e96c03701543912c29921d690b48":{"kind":"REWORK","_number":1,"created":"2023-02-27 13:13:47.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/94/875494/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/94/875494/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/94/875494/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/94/875494/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/94/875494/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/94/875494/1"}}},"commit":{"parents":[{"commit":"8726503615a9a23038174bcae749378febec93c1","subject":"Replace deprecated defined resource type of Manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8726503615a9a23038174bcae749378febec93c1"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2023-02-27 13:06:40.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2023-02-27 13:13:45.000000000","tz":60},"subject":"Enable snapshots in NFS backends","message":"Enable snapshots in NFS backends\n\nThis feature is disabled by default in cinder because of issues with old\nversions of libvirt [1] which are not found in centos8 and centos9, which\nare the versions supported since long time ago.\n\n[1] https://github.com/openstack/cinder/commit/2d77a7a87d625f47deab4955298dad9d21a93f26\n\nChange-Id: Ie70d0f242220b513bab37befedb9ab785e08d077\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ee2b4b668786e96c03701543912c29921d690b48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ee2b4b668786e96c03701543912c29921d690b48"}]},"branch":"refs/heads/master"},"bd1e892301bfbb392c870a314dcbbb404d5a1e99":{"kind":"REWORK","_number":2,"created":"2023-02-27 14:37:36.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/94/875494/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/94/875494/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/94/875494/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/94/875494/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/94/875494/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/94/875494/2"}}},"commit":{"parents":[{"commit":"8726503615a9a23038174bcae749378febec93c1","subject":"Replace deprecated defined resource type of Manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8726503615a9a23038174bcae749378febec93c1"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2023-02-27 13:06:40.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2023-02-27 14:35:25.000000000","tz":60},"subject":"Enable snapshots in NFS backends and add nova section","message":"Enable snapshots in NFS backends and add nova section\n\nSnapshots feature is disabled in nfs by default in cinder because of issues with old\nversions of libvirt [1] which are not found in centos8 and centos9, which\nare the versions supported since long time ago.\n\nAlso, we are missing the [nova] section in cinder.conf which is required\nin some cinder operations, as snapshots with nfs backend.\n\n[1] https://github.com/openstack/cinder/commit/2d77a7a87d625f47deab4955298dad9d21a93f26\n\nChange-Id: Ie70d0f242220b513bab37befedb9ab785e08d077\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bd1e892301bfbb392c870a314dcbbb404d5a1e99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bd1e892301bfbb392c870a314dcbbb404d5a1e99"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}}]}],"submit_requirements":[]}
