)]}'
{"id":"x%2Fpackstack~875952","triplet_id":"x%2Fpackstack~stable%2Fyoga~Ie70d0f242220b513bab37befedb9ab785e08d077","project":"x/packstack","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-12-18 11:46:00.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie70d0f242220b513bab37befedb9ab785e08d077","subject":"Enable snapshots in NFS backends and add nova section","status":"ABANDONED","created":"2023-03-01 18:11:24.000000000","updated":"2024-12-18 11:46:00.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":875494,"cherry_pick_of_patch_set":2,"meta_rev_id":"61bf8d823c441baaa040ca72fd28a45ed54b575f","_number":875952,"virtual_id_number":875952,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-03-02 07:46:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-01 19:17:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f9d85cf02f6639165e1a3c7ff8672059158d8d77","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2023-03-01 18:11:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* packstack/puppet/modules/packstack/manifests/cinder.pp\n* packstack/puppet/modules/packstack/manifests/cinder/backend/nfs.pp","accounts_in_message":[],"_revision_number":1},{"id":"9287016934e58cdd32bd0ed33870444b0e392023","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 19:17:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c2563db228041779a46c456d3297612\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c79aecb604f4d989218cbf682938f2b : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0fa286c0b1a46039531649cff1a742b : SUCCESS in 3m 11s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/0b1cb76ddabd47bfa966187a8ce089ab : FAILURE in 25m 42s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/9908b713918e4529955ca56b89d06367 : SUCCESS in 53m 53s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/7bc6f037e2924ae0ab9a589ec127baa5 : FAILURE in 14m 35s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f94230ca58a34601b788bada9dcf8222 : SUCCESS in 58m 09s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/48452a1b1d3b4c5bb9ddc8bcdad05c3b : FAILURE in 24m 31s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/64773273a8374235ac80908c8c0a3ae0 : SUCCESS in 49m 26s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/87dc2dc2550a405c9b7eb70c1a6bf1af : FAILURE in 10m 40s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ea64c54a5e3a498ea34f16e2abfefd37 : SUCCESS in 45m 08s","accounts_in_message":[],"_revision_number":1},{"id":"86d8d840efb82031a92aab490eca46e32ec77a56","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-03-02 06:21:01.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ed05165e8bc4b143c8326e9da87f3040b95ad9f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 07:46:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4b1189232fa43fda64408ee70cbc842\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7deadfe2a83040328625a984d9f17998 : SUCCESS in 2m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee7b425ec8c2460991815386984e90c0 : SUCCESS in 3m 19s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/6c13081f67834970b731d0746a77c5ee : SUCCESS in 1h 01m 05s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/fdc72d8517b54d5c8ba647ef84e4584a : SUCCESS in 55m 21s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/59e78357915a4965b37ce1b7b8a95bf9 : SUCCESS in 52m 15s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ff5b00d4c3844bc5a041e6ddc022af47 : SUCCESS in 1h 04m 09s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/51787e602deb4c0dbf08f33a029335e6 : SUCCESS in 52m 53s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/25b18f359e7645c696df5a7dd318da31 : SUCCESS in 49m 56s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/806dbc5c7c3a4d1097ef5ac06990fd86 : SUCCESS in 58m 13s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/2c8a959084a84d69a84b0f29ee638bc6 : SUCCESS in 53m 15s","accounts_in_message":[],"_revision_number":2},{"id":"5a39828c8cd7f5665f918f44d325888c62bada9b","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2023-04-11 07:21:43.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"c81f3258a8e8a8eb93268f57548ba77e1d32c210","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2023-04-11 07:22:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a96dcc60d32696a921aa439e9b0431768ad5f255","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 08:33:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/123b629042264271b9ca0f7d80caeb57\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6fc7f40ddd946a0b41a4b59897d352b : SUCCESS in 2m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c32a4744d362430485a7883b79b09432 : SUCCESS in 3m 38s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/e3efe9b4bb594685bce7b790bf26a55e : SUCCESS in 53m 40s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/86125f19833f4324a44c141d77b7278e : SUCCESS in 54m 27s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/024df1528970440ca8b6eb0cfcd0011d : SUCCESS in 59m 01s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/feaaa8ab145f4ef585d3bf799af9cf35 : SUCCESS in 1h 06m 39s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/7a3d53db55a248169476ad70d27902e1 : SUCCESS in 41m 37s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/f42bdf2caf7d44f0bcded8430539d0d5 : SUCCESS in 44m 22s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/3b012ef5bb38415ba0164fd920f177e7 : SUCCESS in 40m 37s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d3fdefe1a7e741e0bd6c813da93c5d07 : SUCCESS in 47m 25s","accounts_in_message":[],"_revision_number":2},{"id":"61bf8d823c441baaa040ca72fd28a45ed54b575f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-12-18 11:46:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"45f244c812d68f47347b910169646584e8eef0f2","revisions":{"7f8d467a6a16016205d31451434106d61e75b9f4":{"kind":"REWORK","_number":1,"created":"2023-03-01 18:11:24.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/52/875952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/52/875952/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/52/875952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/52/875952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/52/875952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/52/875952/1"}}},"commit":{"parents":[{"commit":"8452f2e481ee7c3e111a5271b745497c1149db8e","subject":"Merge \"Declare removal of Python 2 support\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8452f2e481ee7c3e111a5271b745497c1149db8e"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2023-02-27 13:06:40.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2023-03-01 18:11:24.000000000","tz":0},"subject":"Enable snapshots in NFS backends and add nova section","message":"Enable snapshots in NFS backends and add nova section\n\nSnapshots feature is disabled in nfs by default in cinder because of issues with old\nversions of libvirt [1] which are not found in centos8 and centos9, which\nare the versions supported since long time ago.\n\nAlso, we are missing the [nova] section in cinder.conf which is required\nin some cinder operations, as snapshots with nfs backend.\n\n[1] https://github.com/openstack/cinder/commit/2d77a7a87d625f47deab4955298dad9d21a93f26\n\nChange-Id: Ie70d0f242220b513bab37befedb9ab785e08d077\n(cherry picked from commit bd1e892301bfbb392c870a314dcbbb404d5a1e99)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7f8d467a6a16016205d31451434106d61e75b9f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7f8d467a6a16016205d31451434106d61e75b9f4"}]},"branch":"refs/heads/stable/yoga"},"45f244c812d68f47347b910169646584e8eef0f2":{"kind":"REWORK","_number":2,"created":"2023-03-02 06:21:01.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/52/875952/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/52/875952/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/52/875952/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/52/875952/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/52/875952/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/52/875952/2"}}},"commit":{"parents":[{"commit":"8452f2e481ee7c3e111a5271b745497c1149db8e","subject":"Merge \"Declare removal of Python 2 support\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8452f2e481ee7c3e111a5271b745497c1149db8e"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2023-02-27 13:06:40.000000000","tz":60},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2023-03-02 06:21:00.000000000","tz":0},"subject":"Enable snapshots in NFS backends and add nova section","message":"Enable snapshots in NFS backends and add nova section\n\nSnapshots feature is disabled in nfs by default in cinder because of issues with old\nversions of libvirt [1] which are not found in centos8 and centos9, which\nare the versions supported since long time ago.\n\nAlso, we are missing the [nova] section in cinder.conf which is required\nin some cinder operations, as snapshots with nfs backend.\n\n[1] https://github.com/openstack/cinder/commit/2d77a7a87d625f47deab4955298dad9d21a93f26\n\nChange-Id: Ie70d0f242220b513bab37befedb9ab785e08d077\n(cherry picked from commit bd1e892301bfbb392c870a314dcbbb404d5a1e99)\nConflicts:\n  packstack/puppet/modules/packstack/manifests/cinder.pp\n  packstack/puppet/modules/packstack/manifests/cinder/backend/nfs.pp\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/45f244c812d68f47347b910169646584e8eef0f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/45f244c812d68f47347b910169646584e8eef0f2"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
