)]}'
{"id":"x%2Fpackstack~878081","triplet_id":"x%2Fpackstack~master~I8a324fdab89e474ad2a8e6df1560a5c16edbb866","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2023-03-22 10:56:53.000000000","reason":"\u003cGERRIT_ACCOUNT_16312\u003e replied on the change","reason_account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2023-03-22 11:57:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8a324fdab89e474ad2a8e6df1560a5c16edbb866","subject":"Move to project_ terminology used in current Keystone v3 API","status":"MERGED","created":"2023-03-21 14:34:49.000000000","updated":"2023-03-22 11:57:20.000000000","submitted":"2023-03-22 11:57:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878081","meta_rev_id":"f8ebdcb083224ce440c25b8d24dfc52622b051bf","_number":878081,"virtual_id_number":878081,"owner":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-22 11:57:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2023-03-22 10:56:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2023-03-22 10:56:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-21 15:02:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-21 17:16:37.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"62ec2b479512f22212e01dc9abbe561dcd916cbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2023-03-21 14:34:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96b5e88d17b334ec14c74f3a8bb0dace95e9fea4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 15:02:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ecaefdc4f4f415eb88a98c2a782bfd7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65e6df8fb8b14b899f6c6fe8c824b57b : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36409b63366444c396d9dca36373d3a9 : SUCCESS in 4m 10s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/90fb3a38665c497ca4cefc5a3807fcff : FAILURE in 21m 10s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/b764ed662ecc46c08ae7a834d6b13acb : FAILURE in 13m 10s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/169ed30bd21d452bbbd4b5591f39fa86 : FAILURE in 21m 30s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/46b867e7c8204239bb8eafd0aa83f4fb : FAILURE in 22m 48s","accounts_in_message":[],"_revision_number":1},{"id":"50b4391fc81b5022acef3d6aec50341397bf2967","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2023-03-21 15:07:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"da7f5576f5a8cb482c8135ae09080d013d3cda34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 16:15:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a28dd802d2545f886a5be3398948ef6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68d2a15f08bf42ef976bc51a5a7aeffb : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e080d5c389714006900fe0d33f3e7e06 : SUCCESS in 3m 32s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/e9a71a0e89184fbbb112fcc3f4e730e8 : SUCCESS in 42m 24s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/9a4afa324ad0470cb7ce6b287c24c20f : FAILURE in 38m 13s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/c604a98dc4f646578ad7d70c09b64739 : SUCCESS in 1h 00m 07s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/27e953ce32584b15b82ffff9b370aa62 : SUCCESS in 53m 07s","accounts_in_message":[],"_revision_number":2},{"id":"8949706af4470a0d5f880995f27ac9134bb91183","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2023-03-21 16:18:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4fe50066ee4d5038b5b20ea8a7eb3b58fed8bd6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 17:06:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1783dd3b512649b6a83e2d672a163bbc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dace50f49b1b4c3caebd420e41dacd9f : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bab2862cbe5c4e1b8d2a0345661ac760 : SUCCESS in 3m 21s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/e0e84ef07eaf4ad0bcdbaaf5a9125cd5 : SUCCESS in 41m 40s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/11f5ca3733be42fe84b3c9a5f7fea010 : SUCCESS in 43m 13s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/d3102aea1963451a8ff8df74c585ce4d : SUCCESS in 43m 11s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/31285b3c65fa4090bd6ec8f333bf8647 : SUCCESS in 44m 52s","accounts_in_message":[],"_revision_number":2},{"id":"e0cfc43ebca81a731de75010ddd6340caff34230","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2023-03-22 10:56:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0ddf43882ecf37c68705d4f8e9b8a7278a484fdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 10:57:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c12f4b91fbd9d5567bba2a4016e8f9999b376825","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 11:57:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e354446be684a72a55a4c6aa261faae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cdff27c060b40d1904865122ab1bb81 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2767a247c56247958c9a6632407ce95e : SUCCESS in 4m 16s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/bf2a04cb1d5748c2be96e3b05b2563d6 : SUCCESS in 43m 17s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/933a38cdad634ea3bfeb83754b099347 : SUCCESS in 43m 28s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/aa539f5529794a6bb922a11ef7d3e214 : SUCCESS in 59m 39s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/5449abe0b27e448996a047d7911802d0 : SUCCESS in 49m 33s","accounts_in_message":[],"_revision_number":2},{"id":"f8ebdcb083224ce440c25b8d24dfc52622b051bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 11:57:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"15bbe1f9428b6b5be026d60476885b883f43af1a","revisions":{"2d910ae89cb9dc009b880a371b2fc984d536854a":{"kind":"REWORK","_number":1,"created":"2023-03-21 14:34:49.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/81/878081/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/81/878081/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/81/878081/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/81/878081/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/81/878081/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/81/878081/1"}}},"commit":{"parents":[{"commit":"bd1e892301bfbb392c870a314dcbbb404d5a1e99","subject":"Enable snapshots in NFS backends and add nova section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bd1e892301bfbb392c870a314dcbbb404d5a1e99"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2023-03-21 14:30:36.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2023-03-21 14:31:22.000000000","tz":60},"subject":"Move to project_ terminology used in current Keystone v3 API","message":"Move to project_ terminology used in current Keystone v3 API\n\ntenant_* properties were deprecated long time ago with [1]\nand removed recently with [2].\n\n[1] https://review.opendev.org/c/openstack/puppet-neutron/+/828942\n[2] https://review.opendev.org/c/openstack/puppet-neutron/+/877189\n\nChange-Id: I8a324fdab89e474ad2a8e6df1560a5c16edbb866\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2d910ae89cb9dc009b880a371b2fc984d536854a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2d910ae89cb9dc009b880a371b2fc984d536854a"}]},"branch":"refs/heads/master"},"15bbe1f9428b6b5be026d60476885b883f43af1a":{"kind":"REWORK","_number":2,"created":"2023-03-21 15:07:14.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/81/878081/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/81/878081/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/81/878081/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/81/878081/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/81/878081/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/81/878081/2"}}},"commit":{"parents":[{"commit":"bd1e892301bfbb392c870a314dcbbb404d5a1e99","subject":"Enable snapshots in NFS backends and add nova section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bd1e892301bfbb392c870a314dcbbb404d5a1e99"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2023-03-21 14:30:36.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2023-03-21 15:06:51.000000000","tz":60},"subject":"Move to project_ terminology used in current Keystone v3 API","message":"Move to project_ terminology used in current Keystone v3 API\n\ntenant_* properties were deprecated long time ago with [1]\nand removed recently with [2].\n\n[1] https://review.opendev.org/c/openstack/puppet-neutron/+/828942\n[2] https://review.opendev.org/c/openstack/puppet-neutron/+/877189\n\nChange-Id: I8a324fdab89e474ad2a8e6df1560a5c16edbb866\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/15bbe1f9428b6b5be026d60476885b883f43af1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/15bbe1f9428b6b5be026d60476885b883f43af1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
