)]}'
{"id":"x%2Fpackstack~886070","triplet_id":"x%2Fpackstack~master~I60fb8d5306de70c8ec9b813125ec7e11d720e0cf","project":"x/packstack","branch":"master","topic":"ensure_packages","attention_set":{"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2023-06-14 12:10:00.000000000","reason":"\u003cGERRIT_ACCOUNT_31068\u003e replied on the change","reason_account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I60fb8d5306de70c8ec9b813125ec7e11d720e0cf","subject":"Namespace ensure_packages and merge in stdlib puppet module","status":"NEW","created":"2023-06-14 09:09:14.000000000","updated":"2023-10-17 16:19:46.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8121f8c35873cccfcf977f94010a541c31aa2809","_number":886070,"virtual_id_number":886070,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-10-17 16:19:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-17 15:08:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-14 09:57:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-14 12:10:00.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"}],"messages":[{"id":"87e54b0318368c54a449fad7ac103fbeb0b82d9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2023-06-14 09:09:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e4ab1e1bd09c5cdce083ef7b95b71ae41cbe966","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 09:57:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75c60a86193c49f48e80be7ceab450ae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f892788688c640be929d25bb3ab3ba4f : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/587652359080415182eb0709f84fea20 : SUCCESS in 3m 12s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/ed7c0efffb1148e69396b1b4d13202c7 : SUCCESS in 38m 46s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/986d7700ad4042ffad2ce9a3674594fc : SUCCESS in 35m 18s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/9204ee911db14c7da6a92c91f43563e3 : SUCCESS in 41m 07s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/3a8d94e8083e47d9949c58852d9600b6 : SUCCESS in 46m 11s","accounts_in_message":[],"_revision_number":1},{"id":"c53cfd26f7acaf167aafc4d0baf63aac56afbbc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2023-06-14 10:01:53.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"dcb8fb2c702c403dc1dfa7eda4f505e2ac67f8e4","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2023-06-14 12:10:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"29e19c2333cc071aa38f929bc1fad13bacd94ecd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2023-10-17 15:08:42.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"8121f8c35873cccfcf977f94010a541c31aa2809","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 16:19:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/562789d3e2fc4c539d0fc1d6fbaa0a1e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c46dc630aab7428ba88bb24ebc712134 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e67d04794644ea79c2b045692d06f72 : SUCCESS in 3m 31s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/12a1e7051bc345a9b317cd3d22459081 : SUCCESS in 54m 31s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/193fb110a77548ff9765a57277c87806 : SUCCESS in 55m 39s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/e8bf9f49af5344abab7b94102cafd4e0 : SUCCESS in 59m 43s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9ce8969c1d924de49eb01877ec152e99 : SUCCESS in 1h 05m 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f8e9aec4f1b327cba98bc7c6273ee262b77f22e3","revisions":{"bffa6cdb847179868c4dcb6ef15413134b6a1463":{"kind":"REWORK","_number":1,"created":"2023-06-14 09:09:14.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/70/886070/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/70/886070/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/70/886070/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/70/886070/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/70/886070/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/70/886070/1"}}},"commit":{"parents":[{"commit":"f989cf07b653f87b7231a2120b284aae1b284ed7","subject":"Update github host keys","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f989cf07b653f87b7231a2120b284aae1b284ed7"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2023-06-14 09:05:42.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2023-06-14 09:09:04.000000000","tz":120},"subject":"Namespace ensure_packages in stdlib puppet module","message":"Namespace ensure_packages in stdlib puppet module\n\nRequired after [1] as documented in [2].\n\n[1] https://github.com/puppetlabs/puppetlabs-stdlib/pull/1356/\n[2] https://github.com/puppetlabs/puppetlabs-stdlib/blob/24a0f2138f0fa3ef535b3424c314324ff3955680/REFERENCE.md#functions\n\nChange-Id: I60fb8d5306de70c8ec9b813125ec7e11d720e0cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bffa6cdb847179868c4dcb6ef15413134b6a1463"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bffa6cdb847179868c4dcb6ef15413134b6a1463"}]},"branch":"refs/heads/master"},"f70d66db2671e9e5be0045718cd35ca92689e93d":{"kind":"REWORK","_number":2,"created":"2023-06-14 10:01:53.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/70/886070/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/70/886070/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/70/886070/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/70/886070/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/70/886070/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/70/886070/2"}}},"commit":{"parents":[{"commit":"f989cf07b653f87b7231a2120b284aae1b284ed7","subject":"Update github host keys","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f989cf07b653f87b7231a2120b284aae1b284ed7"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2023-06-14 09:05:42.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2023-06-14 10:01:31.000000000","tz":120},"subject":"Namespace ensure_packages and merge in stdlib puppet module","message":"Namespace ensure_packages and merge in stdlib puppet module\n\nRequired after [1] as documented in [2].\n\n[1] https://github.com/puppetlabs/puppetlabs-stdlib/pull/1356/\n[2] https://github.com/puppetlabs/puppetlabs-stdlib/blob/24a0f2138f0fa3ef535b3424c314324ff3955680/REFERENCE.md#functions\n\nChange-Id: I60fb8d5306de70c8ec9b813125ec7e11d720e0cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f70d66db2671e9e5be0045718cd35ca92689e93d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f70d66db2671e9e5be0045718cd35ca92689e93d"}]},"branch":"refs/heads/master"},"f8e9aec4f1b327cba98bc7c6273ee262b77f22e3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-10-17 15:08:42.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/70/886070/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/70/886070/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/70/886070/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/70/886070/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/70/886070/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/70/886070/3"}}},"commit":{"parents":[{"commit":"b00b07af7e7fd507498190674501ce072f71b16e","subject":"The `action` attribute within the `firewall` type has been removed.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b00b07af7e7fd507498190674501ce072f71b16e"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2023-06-14 09:05:42.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2023-10-17 15:08:42.000000000","tz":0},"subject":"Namespace ensure_packages and merge in stdlib puppet module","message":"Namespace ensure_packages and merge in stdlib puppet module\n\nRequired after [1] as documented in [2].\n\n[1] https://github.com/puppetlabs/puppetlabs-stdlib/pull/1356/\n[2] https://github.com/puppetlabs/puppetlabs-stdlib/blob/24a0f2138f0fa3ef535b3424c314324ff3955680/REFERENCE.md#functions\n\nChange-Id: I60fb8d5306de70c8ec9b813125ec7e11d720e0cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f8e9aec4f1b327cba98bc7c6273ee262b77f22e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f8e9aec4f1b327cba98bc7c6273ee262b77f22e3"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
