)]}'
{"id":"x%2Fpackstack~893707","triplet_id":"x%2Fpackstack~master~I31a04a83e2844fa1ec53a27fccd15fb64cafacb0","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2023-10-05 08:30:41.000000000","reason":"\u003cGERRIT_ACCOUNT_16312\u003e replied on the change","reason_account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2023-10-17 08:19:04.000000000","reason":"\u003cGERRIT_ACCOUNT_31068\u003e replied on the change","reason_account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}},"34750":{"account":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"last_update":"2023-10-17 11:07:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I31a04a83e2844fa1ec53a27fccd15fb64cafacb0","subject":"The `action` attribute within the `firewall` type has been removed.","status":"MERGED","created":"2023-09-05 07:57:49.000000000","updated":"2023-10-17 11:07:58.000000000","submitted":"2023-10-17 11:07:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893707","meta_rev_id":"358b84c44f16b67d04b683e59b258646f2082bfc","_number":893707,"virtual_id_number":893707,"owner":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-17 11:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2023-10-16 15:26:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-17 08:19:04.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2023-10-17 10:03:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-05 08:00:28.000000000","updated_by":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"},{"updated":"2023-09-05 08:00:28.000000000","updated_by":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2023-09-05 08:22:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"600e1ad8f3f524e15406d3fb554a2b9b223d9692","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2023-09-05 07:57:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e9f740de5b800326cb2be0eb3b1f74bdfe9a2f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 08:22:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0c8c0adf7e74f7695bda610951e0bf2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c045c56e87874d38ab99b43634dfac18 : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4130b6956a674aa788cbcd329b8fd3d1 : SUCCESS in 3m 23s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/830da185da674658a3f99d429201983b : FAILURE in 10m 38s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/458c3de1497340beb8064888608ccfe8 : FAILURE in 20m 23s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/046a3aae9b8340d4bdea73afe3bbaf9c : FAILURE in 22m 33s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e77bb3740db446ef9d4503b2feae783e : FAILURE in 16m 03s","accounts_in_message":[],"_revision_number":1},{"id":"c6d1b9bce133641d2ebd199d2c0c089155134af8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2023-09-05 11:13:58.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"706a533438da25253bb2dfd454cf076c39efa777","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 11:46:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3763e2eff85942ae9e8eec42d05b2a1f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2007b8ba509443f906877f65a317242 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b2f7fda6aac4aee89bbf6741dae70a4 : SUCCESS in 3m 13s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/bab33d156b364b07a8eae52c5d1ef393 : FAILURE in 9m 36s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/abb9026625e345468f09fcf040e6b1a0 : FAILURE in 23m 49s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/f832c849069c4e488cd0ad4128f8caba : FAILURE in 25m 42s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f0d8fa4335c3412a8148132a8a6bd4d1 : FAILURE in 26m 01s","accounts_in_message":[],"_revision_number":2},{"id":"183e3e50cc1420e6a9966ef3922f71df645106f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2023-09-06 09:47:33.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3998444f16b7dccdff868f865cd244dca8fdeedf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 10:05:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/479668934980455b87121c9f64c5449b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2d6c7c02aa644fa99361d5f6b4baec8 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1cd843e991e1499aaba9b0542f0d17bf : SUCCESS in 3m 13s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/c536a100a56e4cb2b662e3fe7d3a6665 : FAILURE in 10m 46s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/9a86eb28bd7741fba35a21e97d7e4d45 : FAILURE in 9m 46s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/dd0cb962d3b643469a8a0be95360437a : FAILURE in 10m 31s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/b5847f4541f44184a3e9da0618c6d112 : FAILURE in 12m 35s","accounts_in_message":[],"_revision_number":3},{"id":"411d99f4ac66e07ec6d26f9a433463d06efd24ff","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2023-10-05 08:30:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e21297d2c0ffff257b6cdba97acf00ee57d9dc10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 09:02:01.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d5f21c16b43435c95af55d3940e8d06\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f9a868be6364fb3a3e3c9f7d1c7fd88 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35decb96652c4038aef0287d9265510b : SUCCESS in 3m 24s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/2bac9b87c9db4a6d8c420d21cdad3236 : FAILURE in 21m 24s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/680a49f2c0324177aa3fd8817bdacf6a : FAILURE in 24m 00s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/cadeb44583934614b448ec3aa2139ab9 : FAILURE in 24m 27s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/0d0abef0b5534eea9070696f1e71cbcd : FAILURE in 15m 29s","accounts_in_message":[],"_revision_number":3},{"id":"88a285fde593b506cb2189e42c74f34b13d4495f","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2023-10-13 08:18:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"84937da96044c20231829943947af15d5c79db48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 08:51:04.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a53c226e1324e3c9177c57a6da058f0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01ff967ca327425e8f813e61162fb679 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3839fa1a0c84ec494f5245e6d1a05fb : SUCCESS in 4m 26s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/efe06890ee1042dcaa4633fbab2b77e3 : FAILURE in 26m 33s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/cf3a9eb4ff3f4309b164951bd4763db7 : FAILURE in 22m 07s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/eb007f7311fc40b29b0d26e1e03aa2d6 : FAILURE in 10m 22s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/08c12b9a1db74d979e30d08914bc98ed : FAILURE in 23m 49s","accounts_in_message":[],"_revision_number":3},{"id":"3301fcd2694608e48bfeea692148b92663f5c6e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2023-10-13 13:24:38.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"d3767418d1844519574c30992e83521a22745615","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 13:54:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd11e4c4e75f4c8c9c46e43093514902\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c15be9959af84b6e88d7018f87489c3a : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b96db78ada644762ac2ea85e6466783b : SUCCESS in 4m 12s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/8e684eb5c43649bfaf0b56f5dd01f635 : FAILURE in 22m 28s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/aaeb4b2e5bb54e7b9c6ceafe842dd393 : FAILURE in 23m 04s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a5217eb9fc754f5c8a8deed46800eb52 : FAILURE in 10m 20s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/86239ecd948e43aba0ab865407de242c : FAILURE in 24m 36s","accounts_in_message":[],"_revision_number":4},{"id":"a4b4ec4547c698cf8475c7ccf211bdf968e38af0","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2023-10-16 13:04:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ef9ea8ff56518660210a4004f0d6e0965e6c6fd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2023-10-16 13:16:49.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a8d2d78c302c89ec7caf3f041325db8eb7c91840","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 14:01:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45244c69bb204ca8bb7d343aadf708ac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84d779734b34456fb19b49d69d1faac9 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59e19b7dbd2240d686cbd24fd8c1186c : SUCCESS in 6m 11s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/980bc3aee3ae4525b9ee85f42c88ce0a : SUCCESS in 31m 39s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/eaa544fd705a49118c22b91501705e86 : SUCCESS in 37m 58s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a35c1814a6cd4366af0662362a6b0026 : SUCCESS in 41m 38s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/bdd266598121419d9dca91f060cd58cf : SUCCESS in 43m 24s","accounts_in_message":[],"_revision_number":5},{"id":"0de2c36c499b4b499865b7d4e6d128985c914b83","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2023-10-16 15:26:33.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"6a3324f5763942e20bb90f2acbb78259922d871e","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2023-10-17 08:19:04.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"0c5615820076d6a805ffb2b13ba338411feec363","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2023-10-17 10:03:10.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"2e79e97aac82cefe07432430152c34b6aa84e3a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 10:03:53.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"40dc71e1894ba56c7918ca78fd3c2973a21a6a59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 11:07:58.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a379e2f77d904f4c81be488397ff8b98\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7365b081cbdb4ff0b9c40c9b95d33e61 : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6a3295623fb44ec81d06c78ac958a4a : SUCCESS in 4m 16s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/f168061269214d70b58328dd1e88d457 : SUCCESS in 50m 48s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/489604774c9b43c1a826b4d50e614481 : SUCCESS in 29m 58s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/25828a1e3cd248679f0cf27ed6417065 : SUCCESS in 54m 17s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9774bd18b90c43c699300c662a20388a : SUCCESS in 59m 25s","accounts_in_message":[],"_revision_number":5},{"id":"358b84c44f16b67d04b683e59b258646f2082bfc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 11:07:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b00b07af7e7fd507498190674501ce072f71b16e","revisions":{"04aa70bd30eae5d0dd666c7967c424da2e5434af":{"kind":"REWORK","_number":1,"created":"2023-09-05 07:57:49.000000000","uploader":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"ref":"refs/changes/07/893707/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/07/893707/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/07/893707/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/07/893707/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/07/893707/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/07/893707/1"}}},"commit":{"parents":[{"commit":"f989cf07b653f87b7231a2120b284aae1b284ed7","subject":"Update github host keys","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f989cf07b653f87b7231a2120b284aae1b284ed7"}]}],"author":{"name":"karolinku","email":"kkula@redhat.com","date":"2023-09-05 07:50:13.000000000","tz":120},"committer":{"name":"karolinku","email":"kkula@redhat.com","date":"2023-09-05 07:57:46.000000000","tz":120},"subject":" The `action` attribute within the `firewall` type has been removed.","message":" The `action` attribute within the `firewall` type has been removed.\n\nv7.0.0 of firwall module is introducing several changes, as described\nin [1]. This commit is adjusting manifests to this change.\n\n[1] https://github.com/puppetlabs/puppetlabs-firewall/commit/8ea5f45d0040675815f4c0d51fd6716ab5cbe896\n\nChange-Id: I31a04a83e2844fa1ec53a27fccd15fb64cafacb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/04aa70bd30eae5d0dd666c7967c424da2e5434af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/04aa70bd30eae5d0dd666c7967c424da2e5434af"}]},"branch":"refs/heads/master"},"b8b4d82f9630e4cd2e71dd5c1c1f937fed6dbb9a":{"kind":"REWORK","_number":2,"created":"2023-09-05 11:13:58.000000000","uploader":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"ref":"refs/changes/07/893707/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/07/893707/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/07/893707/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/07/893707/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/07/893707/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/07/893707/2"}}},"commit":{"parents":[{"commit":"f989cf07b653f87b7231a2120b284aae1b284ed7","subject":"Update github host keys","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f989cf07b653f87b7231a2120b284aae1b284ed7"}]}],"author":{"name":"karolinku","email":"kkula@redhat.com","date":"2023-09-05 07:50:13.000000000","tz":120},"committer":{"name":"karolinku","email":"kkula@redhat.com","date":"2023-09-05 11:13:47.000000000","tz":120},"subject":" The `action` attribute within the `firewall` type has been removed.","message":" The `action` attribute within the `firewall` type has been removed.\n\nv7.0.0 of firwall module is introducing several changes, as described\nin [1]. This commit is adjusting manifests to this change.\n\n[1] https://github.com/puppetlabs/puppetlabs-firewall/commit/8ea5f45d0040675815f4c0d51fd6716ab5cbe896\n\nChange-Id: I31a04a83e2844fa1ec53a27fccd15fb64cafacb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b8b4d82f9630e4cd2e71dd5c1c1f937fed6dbb9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b8b4d82f9630e4cd2e71dd5c1c1f937fed6dbb9a"}]},"branch":"refs/heads/master"},"77ba3ed493e333a9853687a279021f3720526874":{"kind":"REWORK","_number":3,"created":"2023-09-06 09:47:33.000000000","uploader":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"ref":"refs/changes/07/893707/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/07/893707/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/07/893707/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/07/893707/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/07/893707/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/07/893707/3"}}},"commit":{"parents":[{"commit":"f989cf07b653f87b7231a2120b284aae1b284ed7","subject":"Update github host keys","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f989cf07b653f87b7231a2120b284aae1b284ed7"}]}],"author":{"name":"karolinku","email":"kkula@redhat.com","date":"2023-09-05 07:50:13.000000000","tz":120},"committer":{"name":"karolinku","email":"kkula@redhat.com","date":"2023-09-06 09:45:34.000000000","tz":120},"subject":" The `action` attribute within the `firewall` type has been removed.","message":" The `action` attribute within the `firewall` type has been removed.\n\nv7.0.0 of firwall module is introducing several changes, as described\nin [1]. This commit is adjusting manifests to this change.\n\n[1] https://github.com/puppetlabs/puppetlabs-firewall/commit/8ea5f45d0040675815f4c0d51fd6716ab5cbe896\n\nChange-Id: I31a04a83e2844fa1ec53a27fccd15fb64cafacb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/77ba3ed493e333a9853687a279021f3720526874"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/77ba3ed493e333a9853687a279021f3720526874"}]},"branch":"refs/heads/master"},"54938ff6a0877a178aed44efbf1a4c99216061e1":{"kind":"REWORK","_number":4,"created":"2023-10-13 13:24:38.000000000","uploader":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"ref":"refs/changes/07/893707/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/07/893707/4","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/07/893707/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/07/893707/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/07/893707/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/07/893707/4"}}},"commit":{"parents":[{"commit":"f989cf07b653f87b7231a2120b284aae1b284ed7","subject":"Update github host keys","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f989cf07b653f87b7231a2120b284aae1b284ed7"}]}],"author":{"name":"karolinku","email":"kkula@redhat.com","date":"2023-09-05 07:50:13.000000000","tz":120},"committer":{"name":"karolinku","email":"kkula@redhat.com","date":"2023-10-13 13:24:27.000000000","tz":120},"subject":"DNM  The `action` attribute within the `firewall` type has been removed.","message":"DNM  The `action` attribute within the `firewall` type has been removed.\n\nv7.0.0 of firwall module is introducing several changes, as described\nin [1]. This commit is adjusting manifests to this change.\n\n[1] https://github.com/puppetlabs/puppetlabs-firewall/commit/8ea5f45d0040675815f4c0d51fd6716ab5cbe896\n\nChange-Id: I31a04a83e2844fa1ec53a27fccd15fb64cafacb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/54938ff6a0877a178aed44efbf1a4c99216061e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/54938ff6a0877a178aed44efbf1a4c99216061e1"}]},"branch":"refs/heads/master"},"b00b07af7e7fd507498190674501ce072f71b16e":{"kind":"REWORK","_number":5,"created":"2023-10-16 13:16:49.000000000","uploader":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"ref":"refs/changes/07/893707/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/07/893707/5","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/07/893707/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/07/893707/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/07/893707/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/07/893707/5"}}},"commit":{"parents":[{"commit":"f989cf07b653f87b7231a2120b284aae1b284ed7","subject":"Update github host keys","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f989cf07b653f87b7231a2120b284aae1b284ed7"}]}],"author":{"name":"karolinku","email":"kkula@redhat.com","date":"2023-09-05 07:50:13.000000000","tz":120},"committer":{"name":"karolinku","email":"kkula@redhat.com","date":"2023-10-16 13:16:35.000000000","tz":120},"subject":"The `action` attribute within the `firewall` type has been removed.","message":"The `action` attribute within the `firewall` type has been removed.\n\nv7.0.0 of firwall module is introducing several changes, as described\nin [1]. This commit is adjusting manifests to this change.\n\n[1] https://github.com/puppetlabs/puppetlabs-firewall/commit/8ea5f45d0040675815f4c0d51fd6716ab5cbe896\n\nChange-Id: I31a04a83e2844fa1ec53a27fccd15fb64cafacb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b00b07af7e7fd507498190674501ce072f71b16e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b00b07af7e7fd507498190674501ce072f71b16e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
