)]}'
{"id":"x%2Fpackstack~893868","triplet_id":"x%2Fpackstack~master~Ib7c9cf2af80cdd965b23648d129ecdc599df88c7","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"34750":{"account":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"last_update":"2024-12-24 09:59:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib7c9cf2af80cdd965b23648d129ecdc599df88c7","subject":"DNM test pipeline","status":"ABANDONED","created":"2023-09-06 12:18:06.000000000","updated":"2024-12-24 09:59:01.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b531cd305dacc9fb3e09f634a9c35911f05ce8ae","_number":893868,"virtual_id_number":893868,"owner":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-06 12:47:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-06 12:47:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f35bb708185f38d976ee03abf43d7b9f3f9dcc5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2023-09-06 12:18:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d730150bd1f402d3dfc5556d66350f64a1d7a6fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 12:47:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/62de00b2502f4a34beb9f91cac7c7885\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4f010697e97c49e3bf4a4ca8845d4e0f : SUCCESS in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbe524b13fee47f78f3e1e5c89f6d8e1 : SUCCESS in 4m 14s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/fd2bd9b707194852b5039202c15a332a : FAILURE in 24m 20s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/cea37d366a8b421dbadbb101b7ea04c6 : FAILURE in 23m 05s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/472ee9b3fc774fa7b6091ed058e54b6d : FAILURE in 22m 12s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/68ec142a5d9745aa85ebe017228332fb : FAILURE in 24m 18s","accounts_in_message":[],"_revision_number":1},{"id":"2b7e43247499509d739d1a1d94fde42c6f69837e","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2023-10-16 08:52:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b993eb46db51e11ded6441251a5180eea8c4173c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 09:19:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2aea5277935f4a1a85d985d108b26ce1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/15fbdb3352fe4de1a200efa76d46c1e5 : SUCCESS in 1m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/039d2711fa474eb7b017b1b41e0d124f : SUCCESS in 2m 43s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/89e7acfb33da47f0a1255152aa6e05dc : FAILURE in 10m 09s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/0853fd36aa024745a9a0c45a0895c845 : FAILURE in 22m 11s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/1e385e7e6d384215a40d1a0ed72582f3 : FAILURE in 9m 23s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d50a63d4cd04425a959ae4a1d023db2f : FAILURE in 11m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b531cd305dacc9fb3e09f634a9c35911f05ce8ae","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-12-24 09:59:01.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff3444455366f80531b26c9590cabf0af22f69a2","revisions":{"ff3444455366f80531b26c9590cabf0af22f69a2":{"kind":"REWORK","_number":1,"created":"2023-09-06 12:18:06.000000000","uploader":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"ref":"refs/changes/68/893868/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/68/893868/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/68/893868/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/68/893868/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/68/893868/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/68/893868/1"}}},"commit":{"parents":[{"commit":"f989cf07b653f87b7231a2120b284aae1b284ed7","subject":"Update github host keys","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f989cf07b653f87b7231a2120b284aae1b284ed7"}]}],"author":{"name":"karolinku","email":"kkula@redhat.com","date":"2023-09-06 12:17:22.000000000","tz":120},"committer":{"name":"karolinku","email":"kkula@redhat.com","date":"2023-09-06 12:17:22.000000000","tz":120},"subject":"DNM test pipeline","message":"DNM test pipeline\n\nChange-Id: Ib7c9cf2af80cdd965b23648d129ecdc599df88c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ff3444455366f80531b26c9590cabf0af22f69a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ff3444455366f80531b26c9590cabf0af22f69a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
