)]}'
{"id":"x%2Fpackstack~904896","triplet_id":"x%2Fpackstack~master~I15dca648cd84ae128171d70a194281eebe32bd8d","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-03-06 19:38:07.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-06 19:38:07.000000000","reason":"Change was submitted"},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2024-03-06 15:54:33.000000000","reason":"\u003cGERRIT_ACCOUNT_31068\u003e replied on the change","reason_account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}},"hashtags":[],"change_id":"I15dca648cd84ae128171d70a194281eebe32bd8d","subject":"Remove sahara support","status":"MERGED","created":"2024-01-07 06:20:21.000000000","updated":"2024-03-06 19:38:07.000000000","submitted":"2024-03-06 19:38:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904896","meta_rev_id":"40c891d2ed64d26d0f5dff1d040f31843d3bd49f","_number":904896,"virtual_id_number":904896,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-06 19:38:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-02-28 15:55:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2024-02-27 10:06:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-03-06 15:54:33.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"date":"2024-02-27 02:40:46.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2024-03-06 18:18:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-07 08:48:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-27 10:06:17.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-02-28 15:55:29.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2024-03-01 08:11:47.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2024-03-01 08:11:47.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"}],"messages":[{"id":"7c3b31153af7243d9969eb49e420d171f0d05211","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-07 06:20:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0cbb98b2d5e311c4dafd751f90ea04df9579a416","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-07 06:20:59.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"8b73aab15c6d4e0c99c838b79b5ba58fdecfe9a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-07 06:21:31.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3bfb23f963b2ca91572ae919c9d3c1acd4d254a0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-07 06:21:52.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"58118647fe1ef396fad75f387b4ac38a88cc845e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-07 06:31:24.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5ee1492b4a20b416b6ffecf7180c7357614e336d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 08:48:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/96a99f30ebe0477ab4925fad679a65b4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7da6faaf780a46379dbff7726b56489d : SUCCESS in 1m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a69ec375e8e04bbaa84ccdbe9b30c333 : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42da5af0538147d1b70b1c35283021c7 : SUCCESS in 3m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/243941a6f30d455f9a0c758a07ec43c9 : SUCCESS in 3m 01s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/5ab087fdcdd9402499bda4780e5714cb : TIMED_OUT in 2h 08m 54s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/d51524a8b63749f5bafb42dde606c17b : TIMED_OUT in 2h 09m 04s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/e3c62acfcc8f4240829081887ffc809c : TIMED_OUT in 2h 08m 53s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/5c90b08ade574f969cf59b0dfe8e052f : POST_FAILURE in 1h 04m 05s","accounts_in_message":[],"_revision_number":3},{"id":"cb6b76a3d3199cc75f455935bd6719dade60c71d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-09 03:05:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"baa9ced2073b13c896af3c067130e25fc23038e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 04:41:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4b8bdeef633441ab89a6bd26f9c1da9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fc920986d089444889d72d7adc55b4dc : SUCCESS in 2m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2c0ae6d82d04eb9a15708af665dc3aa : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f3687f00e4a4affa4f743d54152fc28 : SUCCESS in 3m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c0ea313c7c143e6882fd59cf231d020 : SUCCESS in 3m 56s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/9e59bfeef8f24b8bac5fea3a99faa12c : SUCCESS in 56m 21s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/75687f2d7b0149649ab7d5f23b1c1005 : SUCCESS in 52m 00s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/eed0584af57c4c9bb8709af8d7b943a0 : SUCCESS in 56m 04s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/10c351a7556240fbab0506a3fdc8c3b3 : SUCCESS in 1h 13m 25s","accounts_in_message":[],"_revision_number":3},{"id":"b4a47be46fc22613087b4a8dffc9e9313ba66744","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-11 16:29:10.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"a0ebd15a363019d40534ec99cc72fd2705f2bfd8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-27 02:40:46.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"9abea7c98a448ef3fc45611f4af3151603df74d2","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-02-27 10:06:17.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"4c10d452b077de2c6a2692b6d830fd7d3501098b","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-02-28 15:55:29.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"16bac7ef504e61368d18a57e3813553d87d8b560","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-03-06 15:54:33.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"fccf73f060468ea7630b8a1d8ca9827f9a8026a8","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-03-06 18:18:39.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e5a1f4ab7476c5b99883dca3b6665ac1883660af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 18:19:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bcbbefc97111453da901ea4599ed8487114d13e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 19:38:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a98b174f6fb42aca10d694062f3d742\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4b9c8dbf1b0540039d47c2c34ff015f7 : SUCCESS in 1m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f684032f7cf42738c08b90d55de72e3 : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98b262808e074b27b483e2798a3cc3c4 : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bac6e7ad055c46cf807f4dcda4c9066f : SUCCESS in 2m 42s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/1a967b5e48ec43bd98c3cad861976c93 : SUCCESS in 50m 35s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/bce85087aacd4fe79b177479340c3b24 : SUCCESS in 51m 16s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/92cbd39ce5d24ea9a41456ed77cf95bb : SUCCESS in 1h 09m 31s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/22cbf86d9fe7409682bb731afccb7883 : SUCCESS in 1h 00m 09s","accounts_in_message":[],"_revision_number":3},{"id":"40c891d2ed64d26d0f5dff1d040f31843d3bd49f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 19:38:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"892fec52d828261b9b615b701cd4fcae3ed3e3c3","revisions":{"ae74541c4f72a1226db6b82c95e0e47d45005504":{"kind":"REWORK","_number":1,"created":"2024-01-07 06:20:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/904896/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/96/904896/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/96/904896/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/96/904896/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/96/904896/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/96/904896/1"}}},"commit":{"parents":[{"commit":"f989cf07b653f87b7231a2120b284aae1b284ed7","subject":"Update github host keys","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f989cf07b653f87b7231a2120b284aae1b284ed7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-07 06:18:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-07 06:18:58.000000000","tz":540},"subject":"Remove sahara support","message":"Remove sahara support\n\nThe sahara project was marked inactive[1] during this cycle. Health of\nthe project has not been recovered until the promised deadline (m-2)\nand 2024.1 release won\u0027t be created for this project.\n\nChange-Id: I15dca648cd84ae128171d70a194281eebe32bd8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ae74541c4f72a1226db6b82c95e0e47d45005504"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ae74541c4f72a1226db6b82c95e0e47d45005504"}]},"branch":"refs/heads/master"},"4f09f0526c954be15e6b5f287ea68e44686032ee":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-01-07 06:21:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/904896/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/96/904896/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/96/904896/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/96/904896/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/96/904896/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/96/904896/2"}}},"commit":{"parents":[{"commit":"f989cf07b653f87b7231a2120b284aae1b284ed7","subject":"Update github host keys","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f989cf07b653f87b7231a2120b284aae1b284ed7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-07 06:18:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-07 06:21:18.000000000","tz":540},"subject":"Remove sahara support","message":"Remove sahara support\n\nThe sahara project was marked inactive[1] during this cycle. Health of\nthe project has not been recovered until the promised deadline (m-2)\nand 2024.1 release won\u0027t be created for this project.\n\n[1] https://review.opendev.org/c/openstack/governance/+/899986\n\nChange-Id: I15dca648cd84ae128171d70a194281eebe32bd8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4f09f0526c954be15e6b5f287ea68e44686032ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4f09f0526c954be15e6b5f287ea68e44686032ee"}]},"branch":"refs/heads/master"},"892fec52d828261b9b615b701cd4fcae3ed3e3c3":{"kind":"REWORK","_number":3,"created":"2024-01-07 06:31:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/904896/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/96/904896/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/96/904896/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/96/904896/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/96/904896/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/96/904896/3"}}},"commit":{"parents":[{"commit":"a64a08ede407d700d2882deb5301db0f57b7ea12","subject":"Remove reference to monolithic puppet-openstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a64a08ede407d700d2882deb5301db0f57b7ea12"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-07 06:18:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-07 06:30:43.000000000","tz":540},"subject":"Remove sahara support","message":"Remove sahara support\n\nThe sahara project was marked inactive[1] during this cycle. Health of\nthe project has not been recovered until the promised deadline (m-2)\nand 2024.1 release won\u0027t be created for this project.\n\nChange-Id: I15dca648cd84ae128171d70a194281eebe32bd8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/892fec52d828261b9b615b701cd4fcae3ed3e3c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/892fec52d828261b9b615b701cd4fcae3ed3e3c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
