)]}'
{"id":"x%2Fpackstack~904897","triplet_id":"x%2Fpackstack~master~I42dbc51851b5635fb16b566a88e23bc8eed8fbe7","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-29 09:21:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I42dbc51851b5635fb16b566a88e23bc8eed8fbe7","subject":"Remove reference to monolithic puppet-openstack","status":"MERGED","created":"2024-01-07 06:26:58.000000000","updated":"2024-02-29 09:21:23.000000000","submitted":"2024-02-29 09:21:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904897","meta_rev_id":"f19b9e1fe8cd37e451bc8b4597e851c7468bd46b","_number":904897,"virtual_id_number":904897,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-29 09:21:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2024-02-29 08:15:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2024-02-15 20:25:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2024-02-29 08:15:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-07 08:46:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-15 20:25:46.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-02-29 08:15:15.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"7181bd457e068be0919118155de0960c30e5e69d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-07 06:26:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f69a34e9a396e96105062377deda0f51910c77d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-07 06:28:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ea7ce742d1d7be158a34612bc942ab6b9ea32685","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 08:46:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/006a55f0178b4f29b57a00db34b74c3c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b2b351c11ca4365bc0c21d2d180d842 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01b105df6a8347cf8f7a5420067ad6be : SUCCESS in 4m 22s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/a9eaebd1922e46d5b630b865ac39d7c8 : TIMED_OUT in 2h 09m 07s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/03893c5a8ff84141902a1e8892602f69 : TIMED_OUT in 2h 09m 05s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/6da0517c58d54053a9869fa6d088d5fe : TIMED_OUT in 2h 09m 02s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/716aa7c0aa05409bb6daa0d48c853a7a : POST_FAILURE in 1h 04m 00s","accounts_in_message":[],"_revision_number":2},{"id":"14672bcc59532a869457fbf4afa41d06681b2715","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-09 03:05:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"748033738458d4f12a343676e768dbc26e2b7a70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 04:25:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfa5f123538b4eb3a462bf1faaf948a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a58c97230144497b913aebf595b59c3 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b1ef48d43954317a06977ed8dc28c71 : SUCCESS in 3m 53s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/32b4b523d56845bd9c4c3160616e5ee1 : SUCCESS in 56m 22s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/38333b7c7d4d4872b04f68bc63e2197e : SUCCESS in 52m 07s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/38ce9da0d49f4a75931e1016336c8939 : SUCCESS in 1h 00m 36s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/0de1bc6ea2b74ba7bf64e6dd7339c894 : SUCCESS in 1h 03m 13s","accounts_in_message":[],"_revision_number":2},{"id":"b35a1bd5604b07cbbdc52be07327c0cd4f66488e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-02-15 20:25:46.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"df29da4872ab0dfd9d8449f2b9e17eb866056e27","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-02-29 08:15:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8e0bc166bd6494c4181d92ef07b132dcc14f0f85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 08:16:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4e630830d84874e79f1a4bfad90e1f3674bb61c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 09:21:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0de020079dd46fcbd32b1f16cc82654\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2657958e3914088a2824918d674ce2d : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7cc21f2be63b4da7ba38d3935e8eda97 : SUCCESS in 4m 33s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/a7bc9200b492410abd5a42c530c5d669 : SUCCESS in 41m 29s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/e3e400553c58445280c33b5ef42c6f0e : SUCCESS in 51m 01s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/71fbcffd3ed249bf8832825c3a374a12 : SUCCESS in 59m 17s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/740bdbc47024460c9b17e8f9ae1e5472 : SUCCESS in 40m 58s","accounts_in_message":[],"_revision_number":2},{"id":"f19b9e1fe8cd37e451bc8b4597e851c7468bd46b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 09:21:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a64a08ede407d700d2882deb5301db0f57b7ea12","revisions":{"113565de113576c60f779b0f90f2b81b706d35e4":{"kind":"REWORK","_number":1,"created":"2024-01-07 06:26:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/904897/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/97/904897/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/97/904897/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/97/904897/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/97/904897/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/97/904897/1"}}},"commit":{"parents":[{"commit":"f989cf07b653f87b7231a2120b284aae1b284ed7","subject":"Update github host keys","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f989cf07b653f87b7231a2120b284aae1b284ed7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-07 06:25:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-07 06:25:37.000000000","tz":540},"subject":"Remove reference to monolithic puppet-openstack","message":"Remove reference to monolithic puppet-openstack\n\nThe monolithic module[1] was retired in favor of separate modules per\nservice.\n\n[1] https://github.com/puppetlabs-toy-chest/puppetlabs-openstack\n\nChange-Id: I42dbc51851b5635fb16b566a88e23bc8eed8fbe7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/113565de113576c60f779b0f90f2b81b706d35e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/113565de113576c60f779b0f90f2b81b706d35e4"}]},"branch":"refs/heads/master"},"a64a08ede407d700d2882deb5301db0f57b7ea12":{"kind":"REWORK","_number":2,"created":"2024-01-07 06:28:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/904897/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/97/904897/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/97/904897/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/97/904897/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/97/904897/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/97/904897/2"}}},"commit":{"parents":[{"commit":"f989cf07b653f87b7231a2120b284aae1b284ed7","subject":"Update github host keys","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f989cf07b653f87b7231a2120b284aae1b284ed7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-07 06:25:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-07 06:27:46.000000000","tz":540},"subject":"Remove reference to monolithic puppet-openstack","message":"Remove reference to monolithic puppet-openstack\n\nThe monolithic module[1] was retired in favor of separate modules per\nservice.\n\nAlso the puppet-openstack_extras module is not used.\n\n[1] https://github.com/puppetlabs-toy-chest/puppetlabs-openstack\n\nChange-Id: I42dbc51851b5635fb16b566a88e23bc8eed8fbe7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a64a08ede407d700d2882deb5301db0f57b7ea12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a64a08ede407d700d2882deb5301db0f57b7ea12"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
