)]}'
{"id":"x%2Fpackstack~904898","triplet_id":"x%2Fpackstack~master~Ide554b21cfb029cd863a348dc13ef5328ac83906","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-29 09:20:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ide554b21cfb029cd863a348dc13ef5328ac83906","subject":"Prohibit retired/unmaintained network plugins","status":"MERGED","created":"2024-01-07 06:38:58.000000000","updated":"2024-02-29 09:20:38.000000000","submitted":"2024-02-29 09:20:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904898","meta_rev_id":"b2c8ee5178f46d478d9499142c4cf9f215189a3d","_number":904898,"virtual_id_number":904898,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-29 09:20:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2024-02-29 08:14:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2024-02-15 20:26:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2024-02-29 08:14:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-07 08:54:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-15 20:26:45.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-02-29 08:14:11.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"f92175e200e4a68a7665d720d32fa5fbccbc3c55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-07 06:38:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4fc634aa67c754e798c0f64decf95892928e4c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-07 06:39:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f6b51d3f779082300f243e044011ed5ab9e3da10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-07 08:54:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/efc492508ec2462abf78be533c29d24c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e50bd398ac264a0cad3fe3e05214f17e : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce5d95c151d84fc2ba36171e5bd0cb85 : SUCCESS in 3m 01s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/c110774dbaff4e98b5cf579451d841d0 : TIMED_OUT in 2h 08m 56s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/cd899a17887740b98abae1affbbe3ac0 : TIMED_OUT in 2h 08m 55s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ea95e144a6634811bbaaf2ac75e20821 : TIMED_OUT in 2h 08m 52s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/fea72c85648d49a8a312ee78e82c8a21 : POST_FAILURE in 1h 00m 49s","accounts_in_message":[],"_revision_number":2},{"id":"48dd7a377eb9fb182df648f3b35c8737e0fff17a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-09 03:05:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a00f95be99ae0415747eb768b21a2b48ee6797b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 04:19:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e69d3cd892040ed9450eb8d2e192d45\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90e42ba651ca4d53b269a5091426a734 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a10964f951e047429aca7e77e3a41a40 : SUCCESS in 3m 43s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/fc5b397fcc114e7890ca7cd5e86bf962 : SUCCESS in 50m 47s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/80fe6ae45e6e43b3adfcaae4629a82f7 : SUCCESS in 44m 10s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/23c72c38d3a244e9b647a097539afc60 : SUCCESS in 43m 31s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/74829c9256934a43b98579cecc0c17ab : SUCCESS in 50m 15s","accounts_in_message":[],"_revision_number":2},{"id":"6759aff20b325510907956a102edd6776d42e3f7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-02-15 20:26:45.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"31e6a2aa55fbb23dc0c2dc4e6969657cde7a68bf","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-02-29 08:14:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9a8b42862bf12d0ef2bf1e1962a9fa4ddefa3d7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 08:14:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"86ff90ec69c1522526b562c6c82092a7354e41e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 09:20:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/248387508eec45ca86e597166ce434a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/648bddc796274eb89e5ec8197f48c064 : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e67cb9487f2f44a093e893e8a2ebca3f : SUCCESS in 3m 41s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/ddf350b3030e45619970c0590d0418db : SUCCESS in 56m 55s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/ab84431b63c544acb0537c5155550390 : SUCCESS in 55m 40s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/8d41b4467d434c78b703b0253a637b78 : SUCCESS in 51m 51s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9d6f8d44d9464401bae1681571c304fa : SUCCESS in 1h 01m 07s","accounts_in_message":[],"_revision_number":2},{"id":"b2c8ee5178f46d478d9499142c4cf9f215189a3d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 09:20:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"baf74924e732f71c7302e1400eab500eabd4f3c2","revisions":{"5e7d60135baec321453b7ff40f37352b6013bb99":{"kind":"REWORK","_number":1,"created":"2024-01-07 06:38:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/904898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/98/904898/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/98/904898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/98/904898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/98/904898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/98/904898/1"}}},"commit":{"parents":[{"commit":"f989cf07b653f87b7231a2120b284aae1b284ed7","subject":"Update github host keys","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f989cf07b653f87b7231a2120b284aae1b284ed7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-07 06:34:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-07 06:38:16.000000000","tz":540},"subject":"Prohibit retired/unmaintained network plugins","message":"Prohibit retired/unmaintained network plugins\n\n- The networking-hyperv plugin[1] was retired because the Winstackers\n  project was retired.\n\n- The ncs plugin was already removed[2]\n\n- The cisco_nexues plugin still exists but the networking-cisco project\n  is no longer maintained and has not been updated for 5 years[3].\n\n[1] https://github.com/openstack-archive/networking-hyperv\n[2] https://opendev.org/x/networking-cisco/commit/31e4880299d04ceb399aa38097fc5f2b26e30ab1\n[3] https://opendev.org/x/networking-cisco\n\nChange-Id: Ide554b21cfb029cd863a348dc13ef5328ac83906\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5e7d60135baec321453b7ff40f37352b6013bb99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5e7d60135baec321453b7ff40f37352b6013bb99"}]},"branch":"refs/heads/master"},"baf74924e732f71c7302e1400eab500eabd4f3c2":{"kind":"REWORK","_number":2,"created":"2024-01-07 06:39:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/904898/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/98/904898/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/98/904898/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/98/904898/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/98/904898/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/98/904898/2"}}},"commit":{"parents":[{"commit":"f989cf07b653f87b7231a2120b284aae1b284ed7","subject":"Update github host keys","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f989cf07b653f87b7231a2120b284aae1b284ed7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-07 06:34:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-07 06:39:25.000000000","tz":540},"subject":"Prohibit retired/unmaintained network plugins","message":"Prohibit retired/unmaintained network plugins\n\n- The networking-hyperv plugin[1] was retired because the Winstackers\n  project was retired.\n\n- The ncs plugin was already removed[2]\n\n- The cisco_nexues plugin still exists but the networking-cisco project\n  is no longer maintained and has not been updated for 5 years[3].\n\n[1] https://github.com/openstack-archive/networking-hyperv\n[2] https://opendev.org/x/networking-cisco/commit/31e4880299d04ceb399aa38097fc5f2b26e30ab1\n[3] https://opendev.org/x/networking-cisco\n\nChange-Id: Ide554b21cfb029cd863a348dc13ef5328ac83906\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/baf74924e732f71c7302e1400eab500eabd4f3c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/baf74924e732f71c7302e1400eab500eabd4f3c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
