)]}'
{"id":"x%2Fpackstack~905415","triplet_id":"x%2Fpackstack~stable%2Fzed~I317c5b1fa8967dc40437fa6e3258b7d8f2c79e58","project":"x/packstack","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-01-15 13:27:10.000000000","reason":"Change was submitted"},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2024-01-12 10:42:50.000000000","reason":"\u003cGERRIT_ACCOUNT_16312\u003e replied on the change","reason_account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2024-01-15 13:27:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I317c5b1fa8967dc40437fa6e3258b7d8f2c79e58","subject":"Explicitely disable ssl in mariadb server config","status":"MERGED","created":"2024-01-12 08:18:26.000000000","updated":"2024-01-15 13:27:10.000000000","submitted":"2024-01-15 13:27:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905415","meta_rev_id":"1206fb4381d2b6fe3f740521bb3cda13ade96db9","_number":905415,"virtual_id_number":905415,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-15 13:27:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2024-01-15 12:12:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-12 09:55:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2024-01-15 12:12:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-12 09:21:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-12 09:55:48.000000000","updated_by":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"reviewer":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"state":"REVIEWER"},{"updated":"2024-01-12 10:42:50.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"},{"updated":"2024-01-12 10:42:50.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"1697bc7202fc58427eb4c24143efa2a9faee1e70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-01-12 08:18:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"956458f51a21f09a618784e43f75707a0623316f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 09:21:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b14d7d579fd943b795b24271e40f1c6c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ee1fed3346646cca55b1d78b464bac2 : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25db33f6e0cd4a089ee4f8a258a7a345 : SUCCESS in 3m 10s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/9e3e84fa03da4f9aa1038578e1747523 : SUCCESS in 50m 07s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/22a75da9579546b8bdb074c0bce47fb3 : SUCCESS in 37m 29s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/99edb9c1d7ba40349dc33cf661aea9ce : SUCCESS in 38m 23s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/a8e533f4e49c4bcc836038bd731834ed : SUCCESS in 43m 59s","accounts_in_message":[],"_revision_number":1},{"id":"796b236244e7b6c29c5154f9e9ac1ba45a373c01","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2024-01-12 09:55:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d9c0d5ee823cdbac96ac196cd8d825ae32ce7878","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-01-15 12:12:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f431b2a58b822ccd202d53f9bbaceb7c248ad858","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 12:13:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f62da558af847aebba8a275ef03facd184e0c602","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 13:27:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8428f7853b4e4548a97353e71d80c429\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62504f37097243ac9aee23998bbe50be : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c0126b9535a406ebdcc9ef8a10ccf8b : SUCCESS in 3m 15s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/17a8b75d1ca14fdb9b41a461f80c3c48 : SUCCESS in 40m 07s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/7f22307175844b47810a1182a82b7293 : SUCCESS in 50m 57s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/2da9b9ae57c84c8b94d8432cba68e893 : SUCCESS in 38m 31s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/056a6f0df6cd40c39144fe130e4b2f94 : SUCCESS in 55m 48s","accounts_in_message":[],"_revision_number":1},{"id":"1206fb4381d2b6fe3f740521bb3cda13ade96db9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 13:27:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e41923381e1f4660bb4169d933233b08d5d522ea","revisions":{"e41923381e1f4660bb4169d933233b08d5d522ea":{"kind":"REWORK","_number":1,"created":"2024-01-12 08:18:26.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/15/905415/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/15/905415/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/15/905415/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/15/905415/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/15/905415/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/15/905415/1"}}},"commit":{"parents":[{"commit":"39e908ca1649fb7bb99afbcc767cd9c1c690d19f","subject":"Merge \"Update github host keys\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/39e908ca1649fb7bb99afbcc767cd9c1c690d19f"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-01-12 08:17:07.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-01-12 08:18:24.000000000","tz":60},"subject":"Explicitely disable ssl in mariadb server config","message":"Explicitely disable ssl in mariadb server config\n\nAfter mariadb 10.5.18, mariadb fails to start if ssl cert parameters\nare configured in the file but the certificates do not exist [1]. This\nproblem was fixed in puppetlabs-mysql 13.1.0[2] but we use older\nversions (13.0.0 in stable/zed for example) in stable branch jobs.\n\nCentOS Stream 9 has just bumped to mariadb-10.5.22 and this is breaking\nall non-ssl jobs.\n\n[1] https://mariadb.com/kb/en/mariadb-10-5-18-release-notes/#ssl\n[2] https://github.com/puppetlabs/puppetlabs-mysql/issues/1509\n\nChange-Id: I317c5b1fa8967dc40437fa6e3258b7d8f2c79e58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e41923381e1f4660bb4169d933233b08d5d522ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e41923381e1f4660bb4169d933233b08d5d522ea"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
