)]}'
{"id":"x%2Fpackstack~905446","triplet_id":"x%2Fpackstack~stable%2Fxena~I317c5b1fa8967dc40437fa6e3258b7d8f2c79e58","project":"x/packstack","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2024-01-15 12:12:19.000000000","reason":"\u003cGERRIT_ACCOUNT_16312\u003e replied on the change","reason_account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}},"hashtags":[],"change_id":"I317c5b1fa8967dc40437fa6e3258b7d8f2c79e58","subject":"Explicitely disable ssl in mariadb server config","status":"MERGED","created":"2024-01-12 10:49:06.000000000","updated":"2024-01-15 13:25:15.000000000","submitted":"2024-01-15 13:25:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905446","cherry_pick_of_change":905445,"cherry_pick_of_patch_set":1,"meta_rev_id":"23c0fa067695a4ce720080151e834b1d2dfd279b","_number":905446,"virtual_id_number":905446,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-15 13:25:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2024-01-15 12:12:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2024-01-15 12:12:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-12 11:23:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"36c5d0bed566b3847bb41ed95e680728e88b4e1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-01-12 10:49:06.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"94beeef8e558c8e101b1dab693ff3de41b6ade1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 11:23:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cdb8440deef1441da9e580f3ff26966a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e9fe760e3884931a3b43ea3c8fd0b56 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9101d6a737704d6ead27640c110d2c3a : SUCCESS in 6m 40s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/b9b793e15d584cc99ea6b13fcd1443cf : POST_FAILURE in 13m 39s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/5e1fedf242df4f9b90345675e2d9ffbc : POST_FAILURE in 11m 09s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/effd16172c9a4bdeb656238b091139cf : POST_FAILURE in 10m 50s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ad53e7a847624360aa3427e213d723ae : POST_FAILURE in 29m 09s","accounts_in_message":[],"_revision_number":1},{"id":"477cc2b7186e0ebbef0332ce80cfd63fd45da667","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-01-12 14:35:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6079b9627ebf2d9e718c2cb1dd07dc4608bf7ddc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 15:02:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/45a10f3642fd49cb9fd5fd645ffdd5b8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d86441bf81914df0a41db6da89d1b045 : SUCCESS in 2m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff5f23a460864b0a89491981bee71493 : SUCCESS in 3m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ee8f3de8b5d401e8221c4b1de88ab3c : SUCCESS in 3m 06s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/66cdecb577254c989593bd4c44c0fea6 : POST_FAILURE in 22m 04s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/41fac83343dd46b3bac6b764dbf33ca8 : POST_FAILURE in 9m 28s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/f8eaa9d0dff64db88966652db04489ec : POST_FAILURE in 11m 10s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f35a0e83bc8748e999ddafb84f21f779 : POST_FAILURE in 12m 06s","accounts_in_message":[],"_revision_number":2},{"id":"d0b5f1b4c834384cb7f4ba20bec04939df1bf24a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-01-15 07:31:21.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4a1e9e8448ae133ffd699cd5d257fe35d798bffc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 08:03:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b5c39ba84e340368221585e911b16c0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b48b794e91e4b2bbb87f463aa3cff9a : SUCCESS in 2m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcb5812c241a4f1882299ca3b759b797 : SUCCESS in 3m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0becdf75e1f47efa8eeec356eb0e497 : SUCCESS in 3m 06s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/d120230226c54a39ae554763d2e9d608 : POST_FAILURE in 9m 20s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/afaa651b01534bb399f1559446a277c4 : POST_FAILURE in 9m 06s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/030edfb85196430d84607796c1f11a9a : POST_FAILURE in 11m 52s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f0edf410470c4c9681595aa6ce653efd : POST_FAILURE in 10m 06s","accounts_in_message":[],"_revision_number":3},{"id":"4dbaec2c500f024e8fbcce3c8a5e98a65fc7bc9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-01-15 08:05:58.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a1b49781c6e970eded2b7ac1a0ead06423c7f976","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 09:01:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/58ae1a4596214ca5a1a9cf40226e75c4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9fc1e88c5c84d0f8eebfded53bca48a : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b7b4ada8e7a4257a14d689b7d0e9430 : SUCCESS in 3m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c4f25d8dbc6469e9a4dbf360342c13a : SUCCESS in 3m 54s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/fdc0267f4a9e4d7b87bdc9e7f5305356 : POST_FAILURE in 26m 26s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/04c1d6c8e38545c7ae2aa6c89f2de30f : POST_FAILURE in 12m 43s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/7f2487f682ce4317b045276f5f3023a3 : POST_FAILURE in 13m 29s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/8c1624d13aed4defb5269d967337a4b9 : POST_FAILURE in 27m 05s","accounts_in_message":[],"_revision_number":4},{"id":"87c3707d52b97abeb432fd70c309a0cf7857a27c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-01-15 09:15:26.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"16c24a1f79c6bee1d7c0613294daf3089565fd9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 09:49:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c248b947cd24d8bb01f05edc6ee947b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6fc51a2659248839914240e57dc4976 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b88eb072a62644b799356d92cb486fcd : SUCCESS in 2m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c767a1d10664e77a28e51b686427347 : SUCCESS in 2m 45s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/435718c2eaed4d8393c2586f6b73841a : FAILURE in 27m 55s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/03e688aad5064e05bda096b00caa9317 : FAILURE in 25m 28s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/084558501231462394d53eec9ad78771 : FAILURE in 25m 39s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f27a84c630ac4f8ea118e4df0afcdc47 : FAILURE in 14m 04s","accounts_in_message":[],"_revision_number":5},{"id":"621fd503ad3324eba02b41d1f9c1cbcc21ad4087","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-01-15 10:17:16.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"746d8e1a0a0ccd214c3a3d95a3bd8b08212a853d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 11:47:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abc37613c5dc4d80902eb023e8b38593\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0484488b2b4f4d5494aaea450110b309 : SUCCESS in 2m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/074a5ae49528432fb599cc0b6c2378b7 : SUCCESS in 3m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5e61d692c484bbcac16e02bcb4cd9ee : SUCCESS in 3m 24s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/43adf0618a214311a9d89a63c60c470b : SUCCESS in 53m 44s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/439dc4c71fba4a71aaa8c8e910ac6943 : SUCCESS in 1h 00m 18s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/0615f88f5f7b473f9a0a13a53dd7750f : SUCCESS in 59m 58s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e332f62883104322afe654b55c9b6f1d : SUCCESS in 1h 10m 22s","accounts_in_message":[],"_revision_number":6},{"id":"377550ea40d444a01b612db64e6bb418a9223e36","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-01-15 12:12:19.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"0f1ac7efb523f88a6a8bab07bfb2c9c9734bf918","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 12:12:58.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"51376610b54705c9f5b9eb6991c451d273ae7e4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 13:25:15.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/304393fe76bb4f62a1c3024446795d3a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/022b358e56664740b3675e32b2b1d0b6 : SUCCESS in 2m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a517a5b2b2e94ec6937f6c5a0a7bd3c1 : SUCCESS in 3m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1503f6d4543b4b51960f6d0909ff3480 : SUCCESS in 2m 53s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/8b5e9b0fdaff420395931cb9ea6effa0 : SUCCESS in 54m 51s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/65e41e11dd4c48f5b8590e817ffb0ad7 : SUCCESS in 57m 01s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ef882d8726034281b0f3fb757791ef7e : SUCCESS in 1h 02m 03s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9b346bb49ec24f76bf4d4f34a08f0c84 : SUCCESS in 1h 07m 47s","accounts_in_message":[],"_revision_number":6},{"id":"23c0fa067695a4ce720080151e834b1d2dfd279b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 13:25:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"9d4e9574b63a7c54e7ca97d3fb59d94a3090750c","revisions":{"4244f9958b35babb91f2485e31aec26e6f2544be":{"kind":"REWORK","_number":1,"created":"2024-01-12 10:49:06.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/46/905446/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/46/905446/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/46/905446/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/46/905446/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/46/905446/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/46/905446/1"}}},"commit":{"parents":[{"commit":"b13d33ecb80f85e1b8d95d325c2e93795cd5229e","subject":"Use ecdsa key in CentOS 9 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b13d33ecb80f85e1b8d95d325c2e93795cd5229e"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-01-12 08:17:07.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-01-12 10:49:06.000000000","tz":0},"subject":"Explicitely disable ssl in mariadb server config","message":"Explicitely disable ssl in mariadb server config\n\nAfter mariadb 10.5.18, mariadb fails to start if ssl cert parameters\nare configured in the file but the certificates do not exist [1]. This\nproblem was fixed in puppetlabs-mysql 13.1.0[2] but we use older\nversions (13.0.0 in stable/zed for example) in stable branch jobs.\n\nCentOS Stream 9 has just bumped to mariadb-10.5.22 and this is breaking\nall non-ssl jobs.\n\n[1] https://mariadb.com/kb/en/mariadb-10-5-18-release-notes/#ssl\n[2] https://github.com/puppetlabs/puppetlabs-mysql/issues/1509\n\nChange-Id: I317c5b1fa8967dc40437fa6e3258b7d8f2c79e58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4244f9958b35babb91f2485e31aec26e6f2544be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4244f9958b35babb91f2485e31aec26e6f2544be"}]},"branch":"refs/heads/stable/xena"},"21aeb7f8805955b0744fa85b26a5acd42800d8c1":{"kind":"REWORK","_number":2,"created":"2024-01-12 14:35:02.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/46/905446/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/46/905446/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/46/905446/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/46/905446/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/46/905446/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/46/905446/2"}}},"commit":{"parents":[{"commit":"b13d33ecb80f85e1b8d95d325c2e93795cd5229e","subject":"Use ecdsa key in CentOS 9 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b13d33ecb80f85e1b8d95d325c2e93795cd5229e"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-01-12 08:17:07.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-01-12 14:34:34.000000000","tz":60},"subject":"Explicitely disable ssl in mariadb server config","message":"Explicitely disable ssl in mariadb server config\n\nAfter mariadb 10.5.18, mariadb fails to start if ssl cert parameters\nare configured in the file but the certificates do not exist [1]. This\nproblem was fixed in puppetlabs-mysql 13.1.0[2] but we use older\nversions (13.0.0 in stable/zed for example) in stable branch jobs.\n\nCentOS Stream 9 has just bumped to mariadb-10.5.22 and this is breaking\nall non-ssl jobs.\n\nI\u0027m also fixing u-c url to use the proper version for Xena.\n\n[1] https://mariadb.com/kb/en/mariadb-10-5-18-release-notes/#ssl\n[2] https://github.com/puppetlabs/puppetlabs-mysql/issues/1509\n\nChange-Id: I317c5b1fa8967dc40437fa6e3258b7d8f2c79e58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/21aeb7f8805955b0744fa85b26a5acd42800d8c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/21aeb7f8805955b0744fa85b26a5acd42800d8c1"}]},"branch":"refs/heads/stable/xena"},"099d717a00578656f4c578abf3202235a2046939":{"kind":"REWORK","_number":3,"created":"2024-01-15 07:31:21.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/46/905446/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/46/905446/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/46/905446/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/46/905446/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/46/905446/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/46/905446/3"}}},"commit":{"parents":[{"commit":"b13d33ecb80f85e1b8d95d325c2e93795cd5229e","subject":"Use ecdsa key in CentOS 9 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b13d33ecb80f85e1b8d95d325c2e93795cd5229e"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-01-12 08:17:07.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-01-15 07:31:08.000000000","tz":60},"subject":"Explicitely disable ssl in mariadb server config","message":"Explicitely disable ssl in mariadb server config\n\nAfter mariadb 10.5.18, mariadb fails to start if ssl cert parameters\nare configured in the file but the certificates do not exist [1]. This\nproblem was fixed in puppetlabs-mysql 13.1.0[2] but we use older\nversions (13.0.0 in stable/zed for example) in stable branch jobs.\n\nCentOS Stream 9 has just bumped to mariadb-10.5.22 and this is breaking\nall non-ssl jobs.\n\nI\u0027m also fixing u-c url to use the proper version for Xena.\n\n[1] https://mariadb.com/kb/en/mariadb-10-5-18-release-notes/#ssl\n[2] https://github.com/puppetlabs/puppetlabs-mysql/issues/1509\n\nChange-Id: I317c5b1fa8967dc40437fa6e3258b7d8f2c79e58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/099d717a00578656f4c578abf3202235a2046939"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/099d717a00578656f4c578abf3202235a2046939"}]},"branch":"refs/heads/stable/xena"},"03f04264e7a47100c265cc884c2fd4fca0f2f967":{"kind":"REWORK","_number":4,"created":"2024-01-15 08:05:58.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/46/905446/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/46/905446/4","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/46/905446/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/46/905446/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/46/905446/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/46/905446/4"}}},"commit":{"parents":[{"commit":"b13d33ecb80f85e1b8d95d325c2e93795cd5229e","subject":"Use ecdsa key in CentOS 9 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b13d33ecb80f85e1b8d95d325c2e93795cd5229e"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-01-12 08:17:07.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-01-15 08:04:59.000000000","tz":60},"subject":"Explicitely disable ssl in mariadb server config","message":"Explicitely disable ssl in mariadb server config\n\nAfter mariadb 10.5.18, mariadb fails to start if ssl cert parameters\nare configured in the file but the certificates do not exist [1]. This\nproblem was fixed in puppetlabs-mysql 13.1.0[2] but we use older\nversions (13.0.0 in stable/zed for example) in stable branch jobs.\n\nCentOS Stream 9 has just bumped to mariadb-10.5.22 and this is breaking\nall non-ssl jobs.\n\nI\u0027m also fixing u-c url to use the proper version for Xena and pinning\nRuby gems versions required for CentOS Stream 8.\n\n[1] https://mariadb.com/kb/en/mariadb-10-5-18-release-notes/#ssl\n[2] https://github.com/puppetlabs/puppetlabs-mysql/issues/1509\n\nChange-Id: I317c5b1fa8967dc40437fa6e3258b7d8f2c79e58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/03f04264e7a47100c265cc884c2fd4fca0f2f967"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/03f04264e7a47100c265cc884c2fd4fca0f2f967"}]},"branch":"refs/heads/stable/xena"},"bc1028fec3421d0c88dea9ef0cdd0711d40846df":{"kind":"REWORK","_number":5,"created":"2024-01-15 09:15:26.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/46/905446/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/46/905446/5","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/46/905446/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/46/905446/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/46/905446/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/46/905446/5"}}},"commit":{"parents":[{"commit":"b13d33ecb80f85e1b8d95d325c2e93795cd5229e","subject":"Use ecdsa key in CentOS 9 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b13d33ecb80f85e1b8d95d325c2e93795cd5229e"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-01-12 08:17:07.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-01-15 09:15:18.000000000","tz":60},"subject":"Explicitely disable ssl in mariadb server config","message":"Explicitely disable ssl in mariadb server config\n\nAfter mariadb 10.5.18, mariadb fails to start if ssl cert parameters\nare configured in the file but the certificates do not exist [1]. This\nproblem was fixed in puppetlabs-mysql 13.1.0[2] but we use older\nversions (13.0.0 in stable/zed for example) in stable branch jobs.\n\nCentOS Stream 9 has just bumped to mariadb-10.5.22 and this is breaking\nall non-ssl jobs.\n\nI\u0027m also fixing u-c url to use the proper version for Xena and pinning\nRuby gems versions required for CentOS Stream 8.\n\n[1] https://mariadb.com/kb/en/mariadb-10-5-18-release-notes/#ssl\n[2] https://github.com/puppetlabs/puppetlabs-mysql/issues/1509\n\nChange-Id: I317c5b1fa8967dc40437fa6e3258b7d8f2c79e58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bc1028fec3421d0c88dea9ef0cdd0711d40846df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bc1028fec3421d0c88dea9ef0cdd0711d40846df"}]},"branch":"refs/heads/stable/xena"},"9d4e9574b63a7c54e7ca97d3fb59d94a3090750c":{"kind":"REWORK","_number":6,"created":"2024-01-15 10:17:16.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/46/905446/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/46/905446/6","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/46/905446/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/46/905446/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/46/905446/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/46/905446/6"}}},"commit":{"parents":[{"commit":"b13d33ecb80f85e1b8d95d325c2e93795cd5229e","subject":"Use ecdsa key in CentOS 9 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b13d33ecb80f85e1b8d95d325c2e93795cd5229e"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-01-12 08:17:07.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-01-15 10:17:07.000000000","tz":60},"subject":"Explicitely disable ssl in mariadb server config","message":"Explicitely disable ssl in mariadb server config\n\nAfter mariadb 10.5.18, mariadb fails to start if ssl cert parameters\nare configured in the file but the certificates do not exist [1]. This\nproblem was fixed in puppetlabs-mysql 13.1.0[2] but we use older\nversions (13.0.0 in stable/zed for example) in stable branch jobs.\n\nCentOS Stream 9 has just bumped to mariadb-10.5.22 and this is breaking\nall non-ssl jobs.\n\nI\u0027m also fixing u-c url to use the proper version for Xena and pinning\nRuby gems versions required for CentOS Stream 8.\n\n[1] https://mariadb.com/kb/en/mariadb-10-5-18-release-notes/#ssl\n[2] https://github.com/puppetlabs/puppetlabs-mysql/issues/1509\n\nChange-Id: I317c5b1fa8967dc40437fa6e3258b7d8f2c79e58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d4e9574b63a7c54e7ca97d3fb59d94a3090750c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9d4e9574b63a7c54e7ca97d3fb59d94a3090750c"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
