)]}'
{"id":"x%2Fpackstack~905677","triplet_id":"x%2Fpackstack~stable%2Fyoga~Ic9dbb0a531b5b0afb6c42528a7792bf6233bd2d9","project":"x/packstack","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2024-01-16 14:51:45.000000000","reason":"\u003cGERRIT_ACCOUNT_16312\u003e replied on the change","reason_account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}},"hashtags":[],"change_id":"Ic9dbb0a531b5b0afb6c42528a7792bf6233bd2d9","subject":"Update github host keys in Yoga","status":"MERGED","created":"2024-01-16 10:56:12.000000000","updated":"2024-01-16 16:10:29.000000000","submitted":"2024-01-16 16:10:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905677","meta_rev_id":"800e26beeafd4320e35048bd56de2ea41083685a","_number":905677,"virtual_id_number":905677,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-16 16:10:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2024-01-16 14:51:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2024-01-16 11:00:52.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2024-01-16 14:51:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-16 11:00:52.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"},{"updated":"2024-01-16 12:21:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d2bc2ca877ced29786a8920d17f5b6935cb2cf53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-01-16 10:56:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fafc63640336acd84cc9b191988a5cb6b53ef0a4","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-01-16 11:00:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"70f58607e502c6e131beea6d99cc76136dc21981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 12:21:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0230c113db94deab6e4ccba7906bced\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/051d5dceab7f4a5790614072bb78a7fe : SUCCESS in 2m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a9abeb34a604aa9a626914b325c04b3 : SUCCESS in 2m 55s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/1c45ac10933d45f988434c3a6e56d896 : SUCCESS in 57m 53s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/e4080d3776da476e9895df10dcf545a3 : SUCCESS in 1h 03m 01s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/28b797c87c584c5b9c3d758f2a79d299 : SUCCESS in 1h 18m 17s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/cab07159c80e47e0876f52750dcbcbf2 : SUCCESS in 1h 19m 02s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/8cf0f9cbd14d4b45a06c3574037fa25f : SUCCESS in 56m 31s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/1af1ee13f5354123b05d3f3b3a38a76d : SUCCESS in 41m 03s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/170d54d7e77a472b8bcaf617dc41646a : SUCCESS in 36m 48s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9f036b8b37ed4fa6ac575f9408f4b99f : SUCCESS in 40m 37s","accounts_in_message":[],"_revision_number":1},{"id":"3dc3b393e587f06235565f7120b1a3dd4ad755b8","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-01-16 14:51:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"36385bb9e2bfea105440bc30374f03ce4762f7e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 14:52:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"59f5085e4815283e36857097ebdfb0226e6dbeb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 16:10:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6af25a48ef994d7b926cc50b758cda18\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/852b7ffc1ae34a2b941912486c6d6714 : SUCCESS in 2m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e1f3c48edc246f8aa0959e0df62b852 : SUCCESS in 2m 51s\n- packstack-centos8s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/9ae6ec1821c4483a84b6e30eb15ccc45 : SUCCESS in 1h 06m 45s\n- packstack-centos8s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/ae6f85d8fc5a4bda81bc5878e618ab71 : SUCCESS in 59m 30s\n- packstack-centos8s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/852fff035b374507be1b691598bd0d24 : SUCCESS in 1h 10m 32s\n- packstack-centos8s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/77bbff058cc64dbdac7360f3e972fdba : SUCCESS in 1h 13m 12s","accounts_in_message":[],"_revision_number":1},{"id":"800e26beeafd4320e35048bd56de2ea41083685a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 16:10:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cedde61e9dd80005e443bdcaaf7c76fac7c86b01","revisions":{"cedde61e9dd80005e443bdcaaf7c76fac7c86b01":{"kind":"REWORK","_number":1,"created":"2024-01-16 10:56:12.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/77/905677/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/77/905677/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/77/905677/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/77/905677/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/77/905677/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/77/905677/1"}}},"commit":{"parents":[{"commit":"f304a5bd2f8b1da8c51df13523949a7c6b858a7e","subject":"Explicitely disable ssl in mariadb server config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f304a5bd2f8b1da8c51df13523949a7c6b858a7e"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2023-04-11 10:28:06.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-01-16 10:55:51.000000000","tz":60},"subject":"Update github host keys in Yoga","message":"Update github host keys in Yoga\n\nhost key for github has been updated [1].\n\n[1] https://github.blog/2023-03-23-we-updated-our-rsa-ssh-host-key/\n\nChange-Id: Ic9dbb0a531b5b0afb6c42528a7792bf6233bd2d9\n(cherry picked from commit f989cf07b653f87b7231a2120b284aae1b284ed7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/cedde61e9dd80005e443bdcaaf7c76fac7c86b01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/cedde61e9dd80005e443bdcaaf7c76fac7c86b01"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
