)]}'
{"id":"x%2Fpackstack~906337","triplet_id":"x%2Fpackstack~master~I2e873f82245ff4b1e64f2d3fbb2589e03abda8d9","project":"x/packstack","branch":"master","topic":"rhbz2228733","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-04-16 14:07:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2e873f82245ff4b1e64f2d3fbb2589e03abda8d9","subject":"Nova: Enable service token","status":"MERGED","created":"2024-01-23 05:34:21.000000000","updated":"2024-04-16 14:07:59.000000000","submitted":"2024-04-16 14:07:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906337-rhbz2228733","meta_rev_id":"5b1280918777f2a1aef1cbe5c9df90f63de276c4","_number":906337,"virtual_id_number":906337,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-16 14:07:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2024-04-16 13:05:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2024-04-16 10:32:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2024-04-16 13:05:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-23 06:33:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-23 08:15:02.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-04-16 13:05:38.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"ca064ffee8b4d387d78faa9dbec2bd37a4b9d048","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-23 05:34:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a132e0614131c87ce87bf16b61e91eb6f18dcb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-23 05:35:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"55cf02631e1a1f1d8daee83c8ad96e655689cdc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 06:33:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba1bd8aecb4941db822b6bf4095c97dc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef931ded44604ec0934f1ceb58ca8e1a : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d0d16ddc30343e5b36dbea6d1ae48df : SUCCESS in 3m 17s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/401f5f570e504046a3cdb8d9655eb55f : SUCCESS in 52m 08s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/31afb0e9e21c4af6b8bd327a16a49de4 : SUCCESS in 40m 53s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ef9aa25304a44790b8e7dbcf0563b5db : SUCCESS in 49m 48s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/11597ae320dc418ab5ab8971c677e51f : SUCCESS in 46m 40s","accounts_in_message":[],"_revision_number":2},{"id":"f5d02f39538e75ba51602243720a93b5b647b379","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-01-23 08:15:02.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"635df59ba9f5a0267578eadf5689a6d6a9725d6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-04-16 09:13:54.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":3},{"id":"1b0f97a2ebf8238c2cf0e56461207b5a713b4e91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 10:15:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/010ab862f7314f6fbf6671db075fb66a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34911d22c1e34ba581dc895ad7e09409 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7bb68e60324643b8a85f02b74a05cc2d : SUCCESS in 2m 51s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/a9151a1f7a5e49bfa1f586b534d943d3 : SUCCESS in 54m 19s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/a52398070c054e61becc2f24f8adc763 : SUCCESS in 32m 32s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/95d972a5507b4aa59dade58ae9554d25 : FAILURE in 23m 35s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/3a008a8e54f44c5a9fc6bb9b5fab31db : SUCCESS in 53m 45s","accounts_in_message":[],"_revision_number":3},{"id":"1003ef7b99bcbd50e19f089b1808792d9e802c32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-04-16 10:32:22.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":4},{"id":"fb9017ffec781379ef6b912664f2cd29d1c828a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 11:39:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54ace7b824174a4cbe66ac2e29ab399e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c2e3584afd04b1b8c53724ba8166020 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98dbe034b39e485b91148e1f35b34173 : SUCCESS in 3m 09s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/df24510e009549919fe20a835914ccd8 : SUCCESS in 42m 45s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/c3840c34553c447997a1a6a8c145d1f4 : SUCCESS in 48m 07s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/11cacf3062e0409b9ffd61a04dab5bd6 : SUCCESS in 48m 55s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/703e6c93d5fa45c0b625be9ca47135e6 : SUCCESS in 35m 07s","accounts_in_message":[],"_revision_number":4},{"id":"b8b0ed803a16b8ebe30c260fa007cf16f96e2a14","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-04-16 13:05:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"52853b46621a417379b629580cd01d7a1eacb9c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 13:06:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"413640af286acc614175b0bede0e547b1a091692","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 14:07:59.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23ed779c7aad435ab0168e67a76461a6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48f801cf09d04969acacdcd4a5b5aa44 : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7119d9e11e42474487a7deef7b8ba4f1 : SUCCESS in 2m 55s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/bd4a59f26fd3473ca6e2b92a73d85321 : SUCCESS in 32m 48s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/aa77a92ee9b243459e1ecfca8d880dcb : SUCCESS in 40m 32s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ce89419ff679448dbfe91cae1f516b0c : SUCCESS in 57m 30s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/27f34c21670a4027af892f7a86b19fd5 : SUCCESS in 38m 53s","accounts_in_message":[],"_revision_number":4},{"id":"5b1280918777f2a1aef1cbe5c9df90f63de276c4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 14:07:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4c7254d2820e7e2b44c494e25ac515b37f3128a6","revisions":{"5d91b800b04517b89d0c41b5e774088e9cd2aaa3":{"kind":"REWORK","_number":1,"created":"2024-01-23 05:34:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/37/906337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/37/906337/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/37/906337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/37/906337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/37/906337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/37/906337/1"}}},"commit":{"parents":[{"commit":"f989cf07b653f87b7231a2120b284aae1b284ed7","subject":"Update github host keys","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f989cf07b653f87b7231a2120b284aae1b284ed7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-23 05:33:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-23 05:33:57.000000000","tz":540},"subject":"Nova: Enable service token","message":"Nova: Enable service token\n\nNow service token is required so that nova can update volume\nattachments in cinder.\n\nResolves: rhbz#2228733\nChange-Id: I2e873f82245ff4b1e64f2d3fbb2589e03abda8d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5d91b800b04517b89d0c41b5e774088e9cd2aaa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5d91b800b04517b89d0c41b5e774088e9cd2aaa3"}]},"branch":"refs/heads/master"},"9899f4978a89d0732759be3b7ea333b6f2cab902":{"kind":"REWORK","_number":2,"created":"2024-01-23 05:35:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/37/906337/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/37/906337/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/37/906337/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/37/906337/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/37/906337/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/37/906337/2"}}},"commit":{"parents":[{"commit":"f989cf07b653f87b7231a2120b284aae1b284ed7","subject":"Update github host keys","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f989cf07b653f87b7231a2120b284aae1b284ed7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-23 05:33:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-23 05:35:30.000000000","tz":540},"subject":"Nova: Enable service token","message":"Nova: Enable service token\n\nNow service token is required so that nova can update volume\nattachments in cinder.\n\nResolves: rhbz#2228733\nChange-Id: I2e873f82245ff4b1e64f2d3fbb2589e03abda8d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9899f4978a89d0732759be3b7ea333b6f2cab902"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/9899f4978a89d0732759be3b7ea333b6f2cab902"}]},"branch":"refs/heads/master"},"953e020ed198a417992212c5ee751193dbf91d9c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-04-16 09:13:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/37/906337/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/37/906337/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/37/906337/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/37/906337/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/37/906337/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/37/906337/3"}}},"commit":{"parents":[{"commit":"6b2c4c9899ab66fbe468c0e536c5f6d60aa9914c","subject":"Generate Puppetfile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6b2c4c9899ab66fbe468c0e536c5f6d60aa9914c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-23 05:33:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-16 09:13:54.000000000","tz":0},"subject":"Nova: Enable service token","message":"Nova: Enable service token\n\nNow service token is required so that nova can update volume\nattachments in cinder.\n\nResolves: rhbz#2228733\nChange-Id: I2e873f82245ff4b1e64f2d3fbb2589e03abda8d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/953e020ed198a417992212c5ee751193dbf91d9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/953e020ed198a417992212c5ee751193dbf91d9c"}]},"branch":"refs/heads/master","description":"Rebase"},"4c7254d2820e7e2b44c494e25ac515b37f3128a6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-04-16 10:32:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/37/906337/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/37/906337/4","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/37/906337/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/37/906337/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/37/906337/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/37/906337/4"}}},"commit":{"parents":[{"commit":"6b60406769ac26967261f89a88cc939ed873b163","subject":"Follow puppet-systemd master again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6b60406769ac26967261f89a88cc939ed873b163"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-23 05:33:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-16 10:32:22.000000000","tz":0},"subject":"Nova: Enable service token","message":"Nova: Enable service token\n\nNow service token is required so that nova can update volume\nattachments in cinder.\n\nResolves: rhbz#2228733\nChange-Id: I2e873f82245ff4b1e64f2d3fbb2589e03abda8d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4c7254d2820e7e2b44c494e25ac515b37f3128a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4c7254d2820e7e2b44c494e25ac515b37f3128a6"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
