)]}'
{"id":"x%2Fpackstack~912479","triplet_id":"x%2Fpackstack~master~Ic1e59df1cd86e2d3ad728236a34e4ea57a2ff89a","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2024-03-21 14:01:28.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic1e59df1cd86e2d3ad728236a34e4ea57a2ff89a","subject":"Remove pupept-ssh requirement","status":"ABANDONED","created":"2024-03-11 17:14:19.000000000","updated":"2024-03-21 14:01:28.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"383ab0a7a874f4a0de27fbf813054d4af40bf572","_number":912479,"virtual_id_number":912479,"owner":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-03-12 13:11:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2024-03-12 13:41:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-11 18:19:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-12 13:41:39.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"e24eebf6ff629ce57cf89d99856a77b8573ee003","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-03-11 17:14:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e3baa7baf97d66fb495fdae3e14bbd9bdc6b0a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 18:19:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffe996b5dba745799bea14a93d25cd3e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c45430c2c0fc462bb82c8271d50ee39a : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e23526559894b3c863e0c13db8677be : SUCCESS in 4m 18s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/6541499b220046d28e32649e25934dbd : FAILURE in 23m 24s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/c96fa807bb414687a45f164dff26b5d2 : SUCCESS in 36m 45s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/1ce70f92930d462ebf72f63071edf072 : SUCCESS in 57m 56s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/5bea256cf8234e87ab941742ad0adb41 : SUCCESS in 59m 45s","accounts_in_message":[],"_revision_number":1},{"id":"af2cde054761af240584297b9121bd3f6e9bc28e","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-03-12 08:14:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"42caa08de19dcc64d4644b5232741106a11a2a3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 09:20:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7322d10a38e413baf5d4f45539d8adc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dc510ed821347d6a13292311d0909f3 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c58bcf72e4a4de7815f1dd63065916b : SUCCESS in 3m 15s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/49b8ca89db204b13ab0b1e0a93d26771 : SUCCESS in 36m 34s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/174e4609d6d246538b1e2d1075499628 : FAILURE in 28m 36s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/6e14b665283c45878c99af2c21153fa3 : SUCCESS in 1h 04m 54s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/19a356d92b3c41ccac2a4ed4a14bc159 : SUCCESS in 58m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e73ca9443aeb69dba33b3a41d70cf467d0cacda7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-03-12 10:57:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e0c361acc5abbf48714785879e50c30a2d370405","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-03-12 12:00:02.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"8198822d2a5ed4456e5d378f676b7367fcf979ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 13:11:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f15e789702744003bb0f19a7772d0b8b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfe299fa0d1a4b678330136d0e1c1078 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bacab90d249745f5bac73e92c4045c8d : SUCCESS in 3m 03s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/c8486e55226249c3b6411365a3c3d504 : SUCCESS in 1h 01m 51s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/344271fa933e46b6bef970c975ec39c7 : SUCCESS in 33m 50s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/e5f3d9cd7f7843e68ff5d9b77bd76f48 : SUCCESS in 42m 28s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/38f12cc92713471a941db7cbc003c2f0 : SUCCESS in 1h 00m 58s","accounts_in_message":[],"_revision_number":3},{"id":"f68f4082dc3081dc11985ae8c89d6871dbbda52b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-03-12 13:41:39.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"383ab0a7a874f4a0de27fbf813054d4af40bf572","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-03-21 14:01:28.000000000","message":"Abandoned\n\nI\u0027m abandoning this patch in favor of https://review.opendev.org/c/x/packstack/+/912607 where I\u0027m proposing the removal of ssh as well","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"392459ae585139d6678d3c34c7422d7981f61816","revisions":{"7e4f720490e52fbb01df08d5ee401afca36b1bef":{"kind":"REWORK","_number":1,"created":"2024-03-11 17:14:19.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/79/912479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/79/912479/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/79/912479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/79/912479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/79/912479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/79/912479/1"}}},"commit":{"parents":[{"commit":"6b60406769ac26967261f89a88cc939ed873b163","subject":"Follow puppet-systemd master again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6b60406769ac26967261f89a88cc939ed873b163"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-03-11 17:13:56.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-03-11 17:13:56.000000000","tz":60},"subject":"DNM test w/o ssh","message":"DNM test w/o ssh\n\nChange-Id: Ic1e59df1cd86e2d3ad728236a34e4ea57a2ff89a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7e4f720490e52fbb01df08d5ee401afca36b1bef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7e4f720490e52fbb01df08d5ee401afca36b1bef"}]},"branch":"refs/heads/master"},"eaf8ba5112173d43531ea59456e1083101bd6066":{"kind":"REWORK","_number":2,"created":"2024-03-12 10:57:14.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/79/912479/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/79/912479/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/79/912479/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/79/912479/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/79/912479/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/79/912479/2"}}},"commit":{"parents":[{"commit":"6b60406769ac26967261f89a88cc939ed873b163","subject":"Follow puppet-systemd master again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6b60406769ac26967261f89a88cc939ed873b163"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-03-11 17:13:56.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-03-12 10:57:03.000000000","tz":60},"subject":"DNM test w/o ssh","message":"DNM test w/o ssh\n\nChange-Id: Ic1e59df1cd86e2d3ad728236a34e4ea57a2ff89a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/eaf8ba5112173d43531ea59456e1083101bd6066"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/eaf8ba5112173d43531ea59456e1083101bd6066"}]},"branch":"refs/heads/master"},"392459ae585139d6678d3c34c7422d7981f61816":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-03-12 12:00:02.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/79/912479/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/79/912479/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/79/912479/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/79/912479/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/79/912479/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/79/912479/3"}}},"commit":{"parents":[{"commit":"6b60406769ac26967261f89a88cc939ed873b163","subject":"Follow puppet-systemd master again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6b60406769ac26967261f89a88cc939ed873b163"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-03-11 17:13:56.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-03-12 11:58:13.000000000","tz":60},"subject":"Remove pupept-ssh requirement","message":"Remove pupept-ssh requirement\n\nIt was used by puppet-tripleo [1] which is now retired.\nThere is no other usage of it, so we can remove the\nrequirement.\n\n[1] https://github.com/openstack-archive/puppet-tripleo/blob/18.0.0/manifests/profile/base/sshd.pp\n\nChange-Id: Ic1e59df1cd86e2d3ad728236a34e4ea57a2ff89a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/392459ae585139d6678d3c34c7422d7981f61816"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/392459ae585139d6678d3c34c7422d7981f61816"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
