)]}'
{"id":"x%2Fpackstack~912607","triplet_id":"x%2Fpackstack~master~I2e67c1c7e48b92485066a2e9c1ce4bd5fe602b04","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2024-03-21 14:54:24.000000000","reason":"\u003cGERRIT_ACCOUNT_16312\u003e replied on the change","reason_account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-03-21 16:27:46.000000000","reason":"Change was submitted"},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2024-03-21 16:27:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2e67c1c7e48b92485066a2e9c1ce4bd5fe602b04","subject":"Generate Puppetfile","status":"MERGED","created":"2024-03-12 17:28:49.000000000","updated":"2024-03-21 16:27:46.000000000","submitted":"2024-03-21 16:27:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"912607","meta_rev_id":"5d9e28f89e6d7e5501e2f7b2744733990be2664a","_number":912607,"virtual_id_number":912607,"owner":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-21 16:27:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2024-03-21 14:54:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2024-03-21 14:54:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-12 18:21:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-14 08:39:24.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2024-03-14 11:02:04.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"92a311de89b4585201a7e17f25cedf656a94a031","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-03-12 17:28:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86aa5f5774cbc4c3fc3f8297c634c43186c948d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 18:21:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c46cb7731f694135925a40c082811926\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d382ad2fffbf49b6aa2fadbc936ff71a : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f4a935366d74eb8917764dabc78a85a : SUCCESS in 3m 31s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/4a552adea7184a139f0fb364f51f6d85 : SUCCESS in 35m 39s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/14e5e3a7921d484bb1a1ba468d95ea73 : SUCCESS in 37m 29s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/0e8f69012ef240af85394bf097b39999 : SUCCESS in 39m 26s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/61d15895040449f79afd96025ecf673c : SUCCESS in 45m 18s","accounts_in_message":[],"_revision_number":1},{"id":"c16c2495582593a53e779fec2e8b59626b0cd03f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-03-13 16:14:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0dbecba79ae8ef679012f830b309f219db27d6d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 17:18:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed906571bbd744ec8609e4a34a088c7c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59d26925dea541ab9190b0e8ee63853e : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c92704bce907452c8b65610da961e36a : SUCCESS in 3m 07s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/feceefc7c23f408f8ec7d02baee70d8a : SUCCESS in 58m 28s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/814c7a500ac346da89e4353859df4b1a : SUCCESS in 40m 32s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a3e673c6a0ff4f588064904d47b44961 : SUCCESS in 55m 25s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c10785fcfead4ed3a191d89d63a56f52 : SUCCESS in 42m 12s","accounts_in_message":[],"_revision_number":2},{"id":"53fa830b377e379a3d6082d1760d48919b5a1575","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-03-14 08:38:13.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c96a7c458ea3fd94544c4aac0df207c66167c41d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-03-14 08:38:24.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"1595e6e60a82e12b0c4ca6cf8e336db0f9e41454","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-03-14 08:39:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"24d7fde94b7ff4ce025384d18a5f62fa476b2062","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 09:44:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b762b82e037a4e99b28fdbc5a3d39d1f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18ec77f9b5b145739c5a6e5d6887142b : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f92d8d2eaef49b2add62487abd230c4 : SUCCESS in 3m 53s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/971362d0e302411781824eac7d601566 : SUCCESS in 39m 30s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/e6d8c7423cca483499ad1fc3e228ce1c : SUCCESS in 54m 07s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/12aedc1d043c4457b4a0f464deee04a9 : SUCCESS in 56m 50s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f57e983bbac247bfbf4553a96ff5ee45 : SUCCESS in 1h 01m 21s","accounts_in_message":[],"_revision_number":3},{"id":"412db0f3ed940cf7534961c7193a1935d17500c5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-03-14 11:02:04.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6b09ffa981bfc9d9d8b462f5aec1d40b0ac98f21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-03-21 14:00:28.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"01dec39e0aa187e466d4651512df6ce62406886a","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-03-21 14:54:24.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5276704a10edecf4e2478ca176f5eee80298b7e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 15:14:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eee0314049c4492d884d3f949b24fcd9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba89a9de37e149349fc9279ac5cf3297 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fb360b1d6174d80912a0e158d846e34 : SUCCESS in 3m 02s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/c6d8d63462584432b3c6c868e3d7affc : SUCCESS in 1h 03m 20s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/6fbdb0d31e74473aa4fb886c90abed2a : SUCCESS in 51m 11s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/90abeef720e5412fb5c0cb9482bedfcf : SUCCESS in 59m 17s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/bf375deeb4344203a9441eaea6f371a2 : SUCCESS in 56m 06s","accounts_in_message":[],"_revision_number":4},{"id":"626ef2cabf4f112bd3c45f3887cf8b070493fea9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 15:15:16.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d957f38555e7bf55b47d01ee5928982efdebc1bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 16:27:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d156150787ca4fbfaefe4f693a391b6b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de3af30faaa74a35b150095c11573595 : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e95529ac43fa4498b464130c64e03ad1 : SUCCESS in 3m 28s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/d21f9ff53fe24c86b78e2cb6449e876c : SUCCESS in 58m 54s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/50b9c51199bc49668b1af1cd518c7bc5 : SUCCESS in 55m 15s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/5f9454a44e3c40b28beab0dc75f23e6c : SUCCESS in 43m 47s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/34f522769c12497487544439bce46332 : SUCCESS in 1h 03m 16s","accounts_in_message":[],"_revision_number":4},{"id":"5d9e28f89e6d7e5501e2f7b2744733990be2664a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 16:27:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6b2c4c9899ab66fbe468c0e536c5f6d60aa9914c","revisions":{"f380306550297e710533ac39399f307480c0b741":{"kind":"REWORK","_number":1,"created":"2024-03-12 17:28:49.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/07/912607/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/07/912607/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/07/912607/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/07/912607/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/07/912607/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/07/912607/1"}}},"commit":{"parents":[{"commit":"046c6b4e7143bcfdd77867bdcba8ac1fd00c8941","subject":"Merge \"Add info about bug tracking tool for packstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/046c6b4e7143bcfdd77867bdcba8ac1fd00c8941"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-03-12 17:25:33.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-03-12 17:25:33.000000000","tz":60},"subject":"WIP Generate Puppetfile","message":"WIP Generate Puppetfile\n\nChange-Id: I2e67c1c7e48b92485066a2e9c1ce4bd5fe602b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f380306550297e710533ac39399f307480c0b741"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f380306550297e710533ac39399f307480c0b741"}]},"branch":"refs/heads/master"},"bf316b8662f619b586d1863267f64b7ca02dca51":{"kind":"REWORK","_number":2,"created":"2024-03-13 16:14:18.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/07/912607/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/07/912607/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/07/912607/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/07/912607/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/07/912607/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/07/912607/2"}}},"commit":{"parents":[{"commit":"046c6b4e7143bcfdd77867bdcba8ac1fd00c8941","subject":"Merge \"Add info about bug tracking tool for packstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/046c6b4e7143bcfdd77867bdcba8ac1fd00c8941"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-03-12 17:25:33.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-03-13 16:13:42.000000000","tz":60},"subject":"Generate Puppetfile","message":"Generate Puppetfile\n\nWe want to generate our Puppefile automatically\nbased on the value set in openstack_modules.txt and\nexternal_modules.txt files.\nA bot will then propose the correct version based on\npuppet-openstack-integration project for the common\npuppet modules. For the modules only required by\nPackstack, the bot will propose the latest version.\nNote we can pin by tag or commit if needed.\n\nChange-Id: I2e67c1c7e48b92485066a2e9c1ce4bd5fe602b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bf316b8662f619b586d1863267f64b7ca02dca51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bf316b8662f619b586d1863267f64b7ca02dca51"}]},"branch":"refs/heads/master"},"7887261beab8984944e7c7faa45e777d911c111c":{"kind":"REWORK","_number":3,"created":"2024-03-14 08:38:13.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/07/912607/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/07/912607/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/07/912607/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/07/912607/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/07/912607/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/07/912607/3"}}},"commit":{"parents":[{"commit":"046c6b4e7143bcfdd77867bdcba8ac1fd00c8941","subject":"Merge \"Add info about bug tracking tool for packstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/046c6b4e7143bcfdd77867bdcba8ac1fd00c8941"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-03-12 17:25:33.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-03-14 08:38:05.000000000","tz":60},"subject":"Generate Puppetfile","message":"Generate Puppetfile\n\nWe want to generate our Puppefile automatically\nbased on the value set in openstack_modules.txt and\nexternal_modules.txt files.\nA bot will then propose the correct version based on\npuppet-openstack-integration project for the common\npuppet modules. For the modules only required by\nPackstack, the bot will propose the latest version.\nNote we can pin by tag or commit if needed.\n\nChange-Id: I2e67c1c7e48b92485066a2e9c1ce4bd5fe602b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7887261beab8984944e7c7faa45e777d911c111c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/7887261beab8984944e7c7faa45e777d911c111c"}]},"branch":"refs/heads/master"},"6b2c4c9899ab66fbe468c0e536c5f6d60aa9914c":{"kind":"REWORK","_number":4,"created":"2024-03-21 14:00:28.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/07/912607/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/07/912607/4","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/07/912607/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/07/912607/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/07/912607/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/07/912607/4"}}},"commit":{"parents":[{"commit":"046c6b4e7143bcfdd77867bdcba8ac1fd00c8941","subject":"Merge \"Add info about bug tracking tool for packstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/046c6b4e7143bcfdd77867bdcba8ac1fd00c8941"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-03-12 17:25:33.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2024-03-21 13:50:53.000000000","tz":60},"subject":"Generate Puppetfile","message":"Generate Puppetfile\n\nWe want to generate our Puppefile automatically\nbased on the value set in openstack_modules.txt and\nexternal_modules.txt files.\nA bot will then propose the correct version based on\npuppet-openstack-integration project for the common\npuppet modules. For the modules only required by\nPackstack, the bot will propose the latest version.\nNote we can pin by tag or commit if needed.\n\nSome notes:\n- We move camptocamp/puppet-systemd to voxpupuli namespace\nto avoid the redirection (i.e camptocamp is voxpupuli now)\n- We keep the pinning for nssdb, remote and sysctl as there is\nno update since a while now and we don\u0027t want to get version\nbehing the current commit. We might unpin if there is a new\nversion.\n- We remove usage of puppet-ssh requirement. It was used by\npuppet-tripleo [1] which is now retired. There is no other\nusage of it, so we can remove the requirement.\n- We move from arioch/puppet-redis to voxpupuli/puppet-redis\nwhich is used by p-o-i [2]. Note arioch/puppet-redis has been\ndeprecated since Mar 15th 2019 [3] and author has suggested\nvoxpupuli one as replacement.\n\n[1] https://review.opendev.org/c/x/packstack/+/912479\n[2] https://github.com/openstack/puppet-openstack-integration/blob/b61132232dba85b18b2563480ae21400f4e58774/external_modules.txt#L19\n[3] https://forge.puppet.com/modules/arioch/redis/readme\n\nChange-Id: I2e67c1c7e48b92485066a2e9c1ce4bd5fe602b04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6b2c4c9899ab66fbe468c0e536c5f6d60aa9914c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6b2c4c9899ab66fbe468c0e536c5f6d60aa9914c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
