)]}'
{"id":"x%2Fpackstack~919460","triplet_id":"x%2Fpackstack~master~I1da80c66b0d01cefd5ac1cb89a75cec00e596bb4","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-22 16:09:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1da80c66b0d01cefd5ac1cb89a75cec00e596bb4","subject":"Remove remaining logic for identity v2 api","status":"MERGED","created":"2024-05-13 06:09:19.000000000","updated":"2024-07-22 16:09:02.000000000","submitted":"2024-07-22 16:09:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919460","meta_rev_id":"7e8ec97254d2a2566ca8f126de9980ca15a7b6bb","_number":919460,"virtual_id_number":919460,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-22 16:09:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2024-07-22 15:01:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2024-07-22 13:22:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2024-07-22 15:01:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-13 07:17:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-15 08:36:22.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-07-22 11:22:13.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"ae21e92a3884740a6b56ed1a322babd6c3fe0848","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-13 06:09:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46f7cd2548959e0f5ff992ca62bd45be93366697","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 07:17:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcb6e5be3c58439180be6cc86c055f9f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8b6fd73c3de44a18003907e8cfb38a7 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b34a4d50280d464c938e204b74377772 : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf6d5f3da61b46889c645e1fce44e2cf : SUCCESS in 2m 35s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/31e950c7246940c89200727f04e1a866 : SUCCESS in 47m 36s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/8c7cf110bc4e48c38b4fbf11b0b450f8 : SUCCESS in 52m 46s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/eaed6e58de44491a915955664dea6a16 : SUCCESS in 56m 35s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/98bdbc558b314566883ee8c74135f101 : SUCCESS in 53m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4d81a2cf58b6509e15751cabf9a3ec658254c20b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-05-15 08:36:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cbad3fe97bb9caec46ae72b0c5dda1000aae031b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-22 11:21:54.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":2},{"id":"d04d71fe3a535b32ea42bb5d18902d0a0e800c84","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-22 11:22:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"79fe3e87d3544573ff2eae48aefd4644bbb99ce7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 12:00:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/149c4a13e5d546d2bde456a003983e4f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/136ac5acefdf4eeeb03544f1cf6094bb : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6509bc06f644849ba8e133ad4ef7d89 : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d87ff0620d34aceb99241d1305f6c7f : SUCCESS in 4m 05s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/2b0c0fa56a204c5c9bc3cf57ddcc4b7c : FAILURE in 21m 57s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/7582698a6f484c88be5e42ceebc03158 : FAILURE in 23m 10s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/bdb67740aaed48e892bfa340a6343649 : FAILURE in 19m 41s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/b6a0d9c4865c420782104e14ab1e11b7 : FAILURE in 23m 03s","accounts_in_message":[],"_revision_number":2},{"id":"446959252df869bd45fd94451e11feac8f204ca1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-22 13:22:31.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":3},{"id":"fce517e154bc87ef0618f263b756a8ff68251589","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 14:19:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8698ebab49d64248ad94b227ac89f814\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20864d8151c4405caadfa5a03f40df5b : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1c3f9aeccdc4840b9beaaed118d4ae7 : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d79f5823045446081ab3a33f08abe48 : SUCCESS in 3m 06s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/017909b36073427b9a1b064da6020510 : SUCCESS in 49m 01s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/574f4ed99dad4ac4a46bf86961cf29c4 : SUCCESS in 49m 26s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/b4e08877c8ea4d43be24903a7ac403f6 : SUCCESS in 49m 00s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/bbc12896e7cf4e1abf02287577578018 : SUCCESS in 40m 31s","accounts_in_message":[],"_revision_number":3},{"id":"a5253331f17c0d61dea68b56e7b1e2a7a3c5c1ce","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-22 15:01:49.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"892699131755bad1dbf42bed05f085a4c04a041a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 15:02:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"50c021f6f2a1a598ccc27af5a0c10835587ea07c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 16:09:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fffe1e1db9014a669f01add4a7c7d0a5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9b275ee235b4562a87e986624a43846 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc9dd86759ad4c2bb74e21d1bdbf5d06 : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/255587f5346146da904ce6510d3e51de : SUCCESS in 3m 13s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/f01696e203a54166a74d86de71e34dc3 : SUCCESS in 50m 54s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/7cee181ac0c0478bb41d1832fa4747ab : SUCCESS in 53m 30s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/0e990a88c0b54a43b1d0ad947f18b06f : SUCCESS in 1h 01m 31s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ee71e0bcbbc74eb9a27dc58f164b8acb : SUCCESS in 42m 43s","accounts_in_message":[],"_revision_number":3},{"id":"7e8ec97254d2a2566ca8f126de9980ca15a7b6bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 16:09:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f68fab914412dfed0304d979e5ce4e83b08a4c89","revisions":{"c507493cc62e7cdd31667c033cd1cc1cfc71f868":{"kind":"REWORK","_number":1,"created":"2024-05-13 06:09:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/919460/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/60/919460/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/60/919460/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/60/919460/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/60/919460/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/60/919460/1"}}},"commit":{"parents":[{"commit":"0c138467942556a7e20d3e3b76e020f2e574cee6","subject":"Move to trove::service_credentials","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0c138467942556a7e20d3e3b76e020f2e574cee6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-13 06:07:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-13 06:09:06.000000000","tz":540},"subject":"Remove remaining logic for identity v2 api","message":"Remove remaining logic for identity v2 api\n\nChange-Id: I1da80c66b0d01cefd5ac1cb89a75cec00e596bb4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c507493cc62e7cdd31667c033cd1cc1cfc71f868"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c507493cc62e7cdd31667c033cd1cc1cfc71f868"}]},"branch":"refs/heads/master"},"265a57fe0b5ec8551c71dfbd6a3f725afd3d62d7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-22 11:21:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/60/919460/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/60/919460/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/60/919460/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/60/919460/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/60/919460/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/60/919460/2"}}},"commit":{"parents":[{"commit":"92691dd1b7b9fcde8bca3e7aa65b8d642fbc6850","subject":"Remove redundant spaces from rc files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/92691dd1b7b9fcde8bca3e7aa65b8d642fbc6850"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-13 06:07:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-22 11:21:54.000000000","tz":0},"subject":"Remove remaining logic for identity v2 api","message":"Remove remaining logic for identity v2 api\n\nChange-Id: I1da80c66b0d01cefd5ac1cb89a75cec00e596bb4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/265a57fe0b5ec8551c71dfbd6a3f725afd3d62d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/265a57fe0b5ec8551c71dfbd6a3f725afd3d62d7"}]},"branch":"refs/heads/master","description":"Rebase"},"f68fab914412dfed0304d979e5ce4e83b08a4c89":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-07-22 13:22:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/60/919460/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/60/919460/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/60/919460/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/60/919460/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/60/919460/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/60/919460/3"}}},"commit":{"parents":[{"commit":"99cc91ff932c99590e7a4d13c4c8934b92006462","subject":"Remove usage of ensure_vpnaas_package in neutron::server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/99cc91ff932c99590e7a4d13c4c8934b92006462"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-13 06:07:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-22 13:22:31.000000000","tz":0},"subject":"Remove remaining logic for identity v2 api","message":"Remove remaining logic for identity v2 api\n\nChange-Id: I1da80c66b0d01cefd5ac1cb89a75cec00e596bb4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f68fab914412dfed0304d979e5ce4e83b08a4c89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f68fab914412dfed0304d979e5ce4e83b08a4c89"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
