)]}'
{"id":"x%2Fpackstack~919462","triplet_id":"x%2Fpackstack~master~I820f134688e345a85ebd09ad6adcea2f6c79da12","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-23 08:32:19.000000000","reason":"Change was submitted"},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2024-07-22 15:01:11.000000000","reason":"\u003cGERRIT_ACCOUNT_16312\u003e replied on the change","reason_account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-07-23 08:32:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I820f134688e345a85ebd09ad6adcea2f6c79da12","subject":"Remove logic for old operating systems","status":"MERGED","created":"2024-05-13 06:16:09.000000000","updated":"2024-07-23 08:32:19.000000000","submitted":"2024-07-23 08:32:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919462","meta_rev_id":"ec775389052d43698d47b0a98bec8d0d2dd8b420","_number":919462,"virtual_id_number":919462,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-23 08:32:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2024-07-22 15:01:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2024-07-23 06:17:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-13 07:39:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-15 08:37:15.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-07-22 11:12:43.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"CC"},{"updated":"2024-07-22 15:01:11.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"0cc506a29b4667e87001dfa4de308f6842f6030e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-13 06:16:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a16abe95bf494e7014d7cca7decf651d76f9994","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 07:39:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2df38218f3e41ab988bcab501ba68b6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e11a0fe669214b9a8412226bcff70bc1 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/325d7a0a047849139a9af5c79b46f07b : SUCCESS in 3m 21s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/cd4471a57c9f420e853e14b678ae4e1a : SUCCESS in 41m 31s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/6ad4cb86a9604b68a8939617cea4fe74 : SUCCESS in 38m 50s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/260ffcef78d04e85bfd4f31609b77aef : SUCCESS in 32m 18s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/0014be8d31194de589a73c256e732e3b : SUCCESS in 59m 51s","accounts_in_message":[],"_revision_number":1},{"id":"cedc1b3281f37663c403dea4506cd52b82f1d829","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-05-15 08:37:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2f1a8ac702fcacc6099e417ccae7052a583a6664","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-22 11:12:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"739ce93b2f520b68ffd0c6a03881387828ceba38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-22 13:55:26.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"482e9da2d2363990631f1c50809671d662840b3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-22 13:55:43.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"245328c6a1351b7712dc6c8c952dfe5d8f231bf9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-22 13:55:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0c8bbce7a9f3fdb74ecdbaabbffc177928f5608d","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-22 15:01:11.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f3decc575de3f4315a99e65f9f357198273104fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 15:07:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/399411be65c54b468fdd49a34cdbae36\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc55c3db866344508788a97ca5e42714 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b3c1555c6b54636b5b9f453d73f58ae : SUCCESS in 2m 56s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/4cb4c6fcb9404c54a70e4a5c3e015887 : SUCCESS in 50m 27s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/5d8998c5461d4c279d3ba17d713afdfc : SUCCESS in 37m 32s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/bde280875f74400a996c21457d8d4e5c : SUCCESS in 1h 00m 13s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/3241d446f9f648e1bd9535d9687dd91b : SUCCESS in 40m 00s","accounts_in_message":[],"_revision_number":3},{"id":"cb5bf103ceb1fdf3749d0970de8401d289757a58","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-23 06:15:00.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7569d6ddb375454fdc80b4c0b1b6ea860f6da133","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-23 06:15:35.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"bf13fa9119e618b4e7e238e5f9460c02eee46e5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 06:15:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b02922e265e78466ed178ad2649152c3ae6197a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 06:15:46.000000000","message":"Patch Set 3: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":3},{"id":"65c5ceb7d3b96bfb31b77d914a1e82edc57cdbcd","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-23 06:17:41.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"57c05c94eb07d4ce5cac39f2dbdfb8c171672884","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-23 06:17:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"901d43ca468029f198d4c7f8142a253c5875aea1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 07:26:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d479cb08e479409dbabbb88c82078f0a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43e20b22354e4a16bf81529c4341dac9 : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b77885e256a94159bed9088da671056b : SUCCESS in 2m 52s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/a17dba7e887143af955ec426a13b8ea0 : SUCCESS in 48m 12s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/897ebcb08dcb49f381cf8f5128a5faab : SUCCESS in 35m 41s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/5095c54539104932977c5ef8d8276fa9 : SUCCESS in 59m 26s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/74aacdfd84b7457990f9125ffac86fbe : SUCCESS in 56m 57s","accounts_in_message":[],"_revision_number":3},{"id":"9181b29ac43de6151a68abf552f1c10a92825065","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 07:27:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"af0e03ba56d8a5d6bfbb7091b45a908a7a0c9be9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 08:32:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1a2504cf06946dcb027028c1f129325\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fc8e6971ada42269baa5f072fe50033 : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e12990069b4c48368dd6b8ddd181ad27 : SUCCESS in 3m 16s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/643dd706a3a1464788daafe6eef56847 : SUCCESS in 41m 43s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/b86dc308a9cd4d5c8346ae5e49e9b5e4 : SUCCESS in 54m 33s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/8d85037828c24715b61cbfd3eb65ee0c : SUCCESS in 57m 09s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/6505086b02604550a0c0d8ce539c44f0 : SUCCESS in 46m 42s","accounts_in_message":[],"_revision_number":3},{"id":"ec775389052d43698d47b0a98bec8d0d2dd8b420","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 08:32:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cf7ae3519f2bd5b1cfd14821158c99f6c2bcaeb8","revisions":{"2281369777196da72dcf5a4238852eaa1b96435a":{"kind":"REWORK","_number":1,"created":"2024-05-13 06:16:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/919462/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/62/919462/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/62/919462/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/62/919462/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/62/919462/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/62/919462/1"}}},"commit":{"parents":[{"commit":"0c138467942556a7e20d3e3b76e020f2e574cee6","subject":"Move to trove::service_credentials","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0c138467942556a7e20d3e3b76e020f2e574cee6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-13 06:15:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-13 06:15:38.000000000","tz":540},"subject":"Remove logic for old operating systems","message":"Remove logic for old operating systems\n\nCentOS \u003c\u003d 8 or Fedora \u003c 22 are too old.\n\nChange-Id: I820f134688e345a85ebd09ad6adcea2f6c79da12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2281369777196da72dcf5a4238852eaa1b96435a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/2281369777196da72dcf5a4238852eaa1b96435a"}]},"branch":"refs/heads/master"},"fa17edf1456868c1d127daa94838551b1340a058":{"kind":"REWORK","_number":2,"created":"2024-07-22 13:55:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/919462/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/62/919462/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/62/919462/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/62/919462/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/62/919462/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/62/919462/2"}}},"commit":{"parents":[{"commit":"0c138467942556a7e20d3e3b76e020f2e574cee6","subject":"Move to trove::service_credentials","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0c138467942556a7e20d3e3b76e020f2e574cee6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-13 06:15:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-22 13:55:26.000000000","tz":0},"subject":"Remove logic for old operating systems","message":"Remove logic for old operating systems\n\nCentOS \u003c\u003d 8 or Fedora \u003c 22 are too old.\n\nChange-Id: I820f134688e345a85ebd09ad6adcea2f6c79da12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fa17edf1456868c1d127daa94838551b1340a058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fa17edf1456868c1d127daa94838551b1340a058"}]},"branch":"refs/heads/master"},"cf7ae3519f2bd5b1cfd14821158c99f6c2bcaeb8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-07-22 13:55:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/919462/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/62/919462/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/62/919462/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/62/919462/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/62/919462/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/62/919462/3"}}},"commit":{"parents":[{"commit":"99cc91ff932c99590e7a4d13c4c8934b92006462","subject":"Remove usage of ensure_vpnaas_package in neutron::server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/99cc91ff932c99590e7a4d13c4c8934b92006462"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-13 06:15:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-22 13:55:43.000000000","tz":0},"subject":"Remove logic for old operating systems","message":"Remove logic for old operating systems\n\nCentOS \u003c\u003d 8 or Fedora \u003c 22 are too old.\n\nChange-Id: I820f134688e345a85ebd09ad6adcea2f6c79da12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/cf7ae3519f2bd5b1cfd14821158c99f6c2bcaeb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/cf7ae3519f2bd5b1cfd14821158c99f6c2bcaeb8"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
