)]}'
{"id":"x%2Fpackstack~924042","triplet_id":"x%2Fpackstack~master~I2158e9d8c859daeca9944b2213bdaad22eb2f054","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-22 16:03:33.000000000","reason":"Change was submitted"},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2024-07-22 15:01:25.000000000","reason":"\u003cGERRIT_ACCOUNT_16312\u003e replied on the change","reason_account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}},"hashtags":[],"change_id":"I2158e9d8c859daeca9944b2213bdaad22eb2f054","subject":"Remove redundant spaces from rc files","status":"MERGED","created":"2024-07-12 08:53:55.000000000","updated":"2024-07-22 16:03:33.000000000","submitted":"2024-07-22 16:03:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924042","meta_rev_id":"e0569356998a9ef7da23ed0b8290c7bc7f6df1f1","_number":924042,"virtual_id_number":924042,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-22 16:03:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2024-07-22 15:01:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2024-07-22 15:01:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-12 10:00:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-22 11:13:26.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"93e575b59f3cfc6275e242e08697c08eb467d53a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-12 08:53:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"459b8cc006917f2826860c9c6ea1691b93b4a439","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 10:00:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8bdbc4d531c4004afcd4c347e257ab4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/765697674146425ba3a4b5e2a4efb1c8 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da24028252134a47a61e1ad7d8010442 : SUCCESS in 3m 28s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/c516df0fb6fe4320974248b1c8c678f4 : SUCCESS in 56m 29s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/1986922cee3f40c1bd003e5cf52d32c5 : SUCCESS in 52m 16s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ee253a1334084a5b9b6e3fd830c605ef : SUCCESS in 55m 09s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/60914f10a6d9407b823b6d23c5c348cb : SUCCESS in 57m 42s","accounts_in_message":[],"_revision_number":1},{"id":"04e1cd6fee52d38c9df42103b46eeba2b06384f4","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-22 11:13:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"77efc5c62bc65a05387966d91a8a28870964d717","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 11:14:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a6609650beac557f3c5ef20007433180cecefe08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 11:49:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d43ec2ccba4048cc8c36a9a21979f5ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7920fbc5db047cb8510385378b36c20 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33986cc2581d4837b6cb1142b28b5a03 : SUCCESS in 3m 14s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/d706b73e6b104134ab997d0ad20504ef : FAILURE in 22m 52s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/efb75ff8b2964975b096aec7af8c9e11 : FAILURE in 22m 27s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/cdc90078f53849a7b53b25d9e3630793 : FAILURE in 25m 02s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/042e22e6eb824658a8ebf40533917834 : FAILURE in 14m 30s","accounts_in_message":[],"_revision_number":1},{"id":"a7a7545b069351089a11dca8e8d084fb1b52545a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-22 13:22:50.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":2},{"id":"fb999ecf18aaad5baeab18c21f7667671fb1f521","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 14:24:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac3a7f4a7dda4c498f67f9a371639226\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5ebc7d5b082483bb4b774d263c2071c : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82e284861eb44c6a8f21759bcb66db1b : SUCCESS in 3m 59s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/fdf4a8d04f874ab68465aaef96cab188 : SUCCESS in 41m 03s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/37f9a1c0166240e88875ff30c160024f : SUCCESS in 53m 14s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/0e7516beddde4b37bebdaa4ea7abdf3c : SUCCESS in 56m 40s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/bedbf8ae55c34419bf9347f889537b79 : SUCCESS in 56m 29s","accounts_in_message":[],"_revision_number":2},{"id":"8c96836fe5f7b60f9829d44566273fd87aebb604","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-22 15:01:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ce16ed4613c5801246efefe301d2e3d56d76c092","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 15:02:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9b80525aba74324637d56e8582d079341f010254","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 16:03:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0bbf117f4fe4858940b6bd646c8f841\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a89fecbd2a74d61a2eb911da4802aca : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31089856fbf548849651d2d39296385a : SUCCESS in 3m 21s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/f5b892b907a641a986fa089fcc5703be : SUCCESS in 40m 41s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/ef1bead3016841f7ace58ecfe44bd2c3 : SUCCESS in 46m 34s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/32f907e10aad4868ad14fe49d3464fd7 : SUCCESS in 46m 04s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/75d6b40ec16e40b48754cca1f8d1d561 : SUCCESS in 54m 06s","accounts_in_message":[],"_revision_number":2},{"id":"e0569356998a9ef7da23ed0b8290c7bc7f6df1f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 16:03:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e34380385de0a36c0c6167812d1a6c3f1a5452e4","revisions":{"92691dd1b7b9fcde8bca3e7aa65b8d642fbc6850":{"kind":"REWORK","_number":1,"created":"2024-07-12 08:53:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/924042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/42/924042/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/42/924042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/42/924042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/42/924042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/42/924042/1"}}},"commit":{"parents":[{"commit":"0c138467942556a7e20d3e3b76e020f2e574cee6","subject":"Move to trove::service_credentials","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0c138467942556a7e20d3e3b76e020f2e574cee6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-12 08:52:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-12 08:53:44.000000000","tz":540},"subject":"Remove redundant spaces from rc files","message":"Remove redundant spaces from rc files\n\nChange-Id: I2158e9d8c859daeca9944b2213bdaad22eb2f054\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/92691dd1b7b9fcde8bca3e7aa65b8d642fbc6850"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/92691dd1b7b9fcde8bca3e7aa65b8d642fbc6850"}]},"branch":"refs/heads/master"},"e34380385de0a36c0c6167812d1a6c3f1a5452e4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-22 13:22:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/42/924042/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/42/924042/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/42/924042/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/42/924042/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/42/924042/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/42/924042/2"}}},"commit":{"parents":[{"commit":"99cc91ff932c99590e7a4d13c4c8934b92006462","subject":"Remove usage of ensure_vpnaas_package in neutron::server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/99cc91ff932c99590e7a4d13c4c8934b92006462"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-12 08:52:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-22 13:22:50.000000000","tz":0},"subject":"Remove redundant spaces from rc files","message":"Remove redundant spaces from rc files\n\nChange-Id: I2158e9d8c859daeca9944b2213bdaad22eb2f054\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e34380385de0a36c0c6167812d1a6c3f1a5452e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e34380385de0a36c0c6167812d1a6c3f1a5452e4"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
