)]}'
{"id":"x%2Fpackstack~924600","triplet_id":"x%2Fpackstack~master~I57b5dfbcf1edb260ce556a2201e36e1697fe6c4b","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2024-07-22 14:27:55.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-22 13:22:12.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2024-07-22 14:27:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I57b5dfbcf1edb260ce556a2201e36e1697fe6c4b","subject":"Remove usage of ensure_vpnaas_package in neutron::server","status":"MERGED","created":"2024-07-22 11:18:15.000000000","updated":"2024-07-22 14:27:55.000000000","submitted":"2024-07-22 14:27:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924600","meta_rev_id":"e6d25b7fec7f959f4b2932c15ce076a42e4b648a","_number":924600,"virtual_id_number":924600,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-22 14:27:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-07-22 13:22:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-07-22 13:23:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-22 12:15:51.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-07-22 12:15:51.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"},{"updated":"2024-07-22 12:22:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c4ce3dc604689604c773bfebf3cdadbd487de276","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-22 11:18:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa3fac393b9cab1e973f0f58fa72c6ed997ebc4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 12:22:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2481542708864b1c913b7243eb448aaa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43033f4958004accb6a7573320b6d483 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b7c12e9adcb4d9499716c44504a0659 : SUCCESS in 3m 16s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/fc67e892ea034736a5eac705726299bd : SUCCESS in 56m 33s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/7e870145286e4e58a5927c2e087683c5 : SUCCESS in 44m 25s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/46a4fdebb2d344508cb96400e51c4685 : SUCCESS in 58m 17s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/f8d5d9fafb0346c3bdd955058f83ea56 : SUCCESS in 49m 20s","accounts_in_message":[],"_revision_number":1},{"id":"8006f9f18071b27ead30e3843f2115f32cdfc79a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-22 13:22:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e7e5ff04ecbf79d438092ce5f3f08be496a083a8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-22 13:23:30.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"468faccde1877ca7c6a833cbfbbe97bda2841fa6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 13:24:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41b2066ad213ac566372e8b5bbe3849adb60f05e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 14:27:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0974baa045ad4d0f8e470f2e7ce13987\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/500be396be974e7eb7d00eb6bb7cb7b3 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8451d5d1ea84bb6937e7ac142f5936d : SUCCESS in 6m 14s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/214549df197c45fb87f5512783dd45c2 : SUCCESS in 37m 52s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/89a6e7e20c3544148483f25ad0ec8032 : SUCCESS in 48m 56s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/ede279a6fcd3418692ef17d44503335d : SUCCESS in 59m 09s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/4690f50f94ef43958f4856b19d3ff96f : SUCCESS in 56m 09s","accounts_in_message":[],"_revision_number":1},{"id":"e6d25b7fec7f959f4b2932c15ce076a42e4b648a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 14:27:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"99cc91ff932c99590e7a4d13c4c8934b92006462","revisions":{"99cc91ff932c99590e7a4d13c4c8934b92006462":{"kind":"REWORK","_number":1,"created":"2024-07-22 11:18:15.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/00/924600/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/00/924600/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/00/924600/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/00/924600/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/00/924600/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/00/924600/1"}}},"commit":{"parents":[{"commit":"0c138467942556a7e20d3e3b76e020f2e574cee6","subject":"Move to trove::service_credentials","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0c138467942556a7e20d3e3b76e020f2e574cee6"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-07-22 11:16:49.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-07-22 11:18:13.000000000","tz":120},"subject":"Remove usage of ensure_vpnaas_package in neutron::server","message":"Remove usage of ensure_vpnaas_package in neutron::server\n\nIt has been removed upstream iin [1].\n\n[1] https://review.opendev.org/c/openstack/puppet-neutron/+/923607\n\nChange-Id: I57b5dfbcf1edb260ce556a2201e36e1697fe6c4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/99cc91ff932c99590e7a4d13c4c8934b92006462"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/99cc91ff932c99590e7a4d13c4c8934b92006462"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
