)]}'
{"id":"x%2Fpackstack~924601","triplet_id":"x%2Fpackstack~master~I6a747ee3431a0921e927486ff030732d6e0abed8","project":"x/packstack","branch":"master","attention_set":{"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2024-07-22 12:38:41.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6a747ee3431a0921e927486ff030732d6e0abed8","subject":"Fix VPNaaS deployment and enable it in scenario002","status":"NEW","created":"2024-07-22 11:19:22.000000000","updated":"2024-07-23 10:05:46.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"43b2b97e0dd3dcecec96155ee762d71a2969f060","_number":924601,"virtual_id_number":924601,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-07-23 10:05:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-22 12:38:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d0c23b3c6c8049c8333a96b0d8ab3eac5df9a7a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-22 11:19:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6b3dcee0093ad16f40ddab671a8c770ccd541a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 12:38:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/12e473182f1a454caa6cf464bc63b2d3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/690c1e88928445eba4bbed90d0c19655 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8af680c70a7435b8c554cface0df9be : SUCCESS in 4m 19s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/b1bab22d2f8648848aa47537e0306e75 : SUCCESS in 37m 06s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/f6885e84697c425f8add1c9a96cd1fa3 : FAILURE in 38m 21s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/842d47829f014592932f15242f88294b : SUCCESS in 38m 55s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/12fe9e559d4d476fb57598e82c39cfc7 : FAILURE in 49m 15s","accounts_in_message":[],"_revision_number":1},{"id":"5e6a8951d93170877c04dcf4639b1ee8a836d7f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-23 09:07:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"43b2b97e0dd3dcecec96155ee762d71a2969f060","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 10:05:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6566cca8ea24459cbf635ec805b14283\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8dbca8847dc44a78485d86c7c1f9739 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cce710536ab74e7d84553c463118b366 : SUCCESS in 3m 46s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/08bc3996f58740908c9801ad8671609b : SUCCESS in 44m 10s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/c4391b538e4a4e9ab2356a8bf0c0bdd0 : FAILURE in 45m 39s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/3d055cb3e04348c2bb7123b7b5a463f7 : SUCCESS in 51m 06s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ab02123246994465aa265cf43f4a05f8 : FAILURE in 48m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"819dc1542113750235a2e37b354008779012f829","revisions":{"fdbed02ae086ef3ef0280297cb5476df50309e40":{"kind":"REWORK","_number":1,"created":"2024-07-22 11:19:22.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/01/924601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/01/924601/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/01/924601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/01/924601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/01/924601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/01/924601/1"}}},"commit":{"parents":[{"commit":"99cc91ff932c99590e7a4d13c4c8934b92006462","subject":"Remove usage of ensure_vpnaas_package in neutron::server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/99cc91ff932c99590e7a4d13c4c8934b92006462"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-07-22 11:18:49.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-07-22 11:18:49.000000000","tz":120},"subject":"Enable VPNaaS agent in scenario002","message":"Enable VPNaaS agent in scenario002\n\nTo improve CI coverage.\n\nChange-Id: I6a747ee3431a0921e927486ff030732d6e0abed8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fdbed02ae086ef3ef0280297cb5476df50309e40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/fdbed02ae086ef3ef0280297cb5476df50309e40"}]},"branch":"refs/heads/master"},"819dc1542113750235a2e37b354008779012f829":{"kind":"REWORK","_number":2,"created":"2024-07-23 09:07:19.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/01/924601/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/01/924601/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/01/924601/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/01/924601/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/01/924601/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/01/924601/2"}}},"commit":{"parents":[{"commit":"99cc91ff932c99590e7a4d13c4c8934b92006462","subject":"Remove usage of ensure_vpnaas_package in neutron::server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/99cc91ff932c99590e7a4d13c4c8934b92006462"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-07-22 11:18:49.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-07-23 09:05:38.000000000","tz":120},"subject":"Fix VPNaaS deployment and enable it in scenario002","message":"Fix VPNaaS deployment and enable it in scenario002\n\nservice provider is managed in neutron::service::vpnaas.\n\nChange-Id: I6a747ee3431a0921e927486ff030732d6e0abed8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/819dc1542113750235a2e37b354008779012f829"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/819dc1542113750235a2e37b354008779012f829"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
