)]}'
{"id":"x%2Fpackstack~924699","triplet_id":"x%2Fpackstack~master~Id129a3894f3506862ea139c5221ae8bab029320e","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-23 10:13:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id129a3894f3506862ea139c5221ae8bab029320e","subject":"Strictly require neutron when nova is enabled","status":"MERGED","created":"2024-07-23 07:54:50.000000000","updated":"2024-07-23 10:13:08.000000000","submitted":"2024-07-23 10:13:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924699","meta_rev_id":"de87baaa50d23311073a025d6ceea5a632ced8b5","_number":924699,"virtual_id_number":924699,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-23 10:13:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2024-07-23 09:10:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2024-07-23 09:10:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-23 08:51:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-23 09:10:59.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"a9333dd0f686d50899025c9a5723859c43e76652","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-23 07:54:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3b9e034fc4017a864a72171df8554b2b611fa9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-23 07:55:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"77771fafb3005a1e60ab4e08a197e0005fe38cea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 08:51:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a0d48500d5341cfbb33f52877ef3571\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fea90b2c150d4eb281d7f4c0dac1180d : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f71dd7ea4cfe4a5b9647bff643da1bc2 : SUCCESS in 3m 12s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/da06e98b62bf43b3a08c206790331446 : SUCCESS in 38m 33s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/a5801d4a6b8c4f0198a5941a74e0026d : SUCCESS in 49m 55s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/adaa7da8ae894df1b62efad2ecc9938a : SUCCESS in 43m 05s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/9a05099a30dc491f8dc354c88dfd3918 : SUCCESS in 45m 10s","accounts_in_message":[],"_revision_number":2},{"id":"2b9bb95fc27eef578c4c00d4615a4c1c1be408da","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-23 09:10:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7dd653564033a1c3af46cb066265d2feeb8fe891","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 09:11:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3325e502a620dff070e13dbcf142af482363dc04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 10:13:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db8d3001f30145749df25b0fc6c066ec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b94e9d953514175a0b9e213649ef2f1 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bb6788d2807493b8e7f924b10402ee4 : SUCCESS in 3m 34s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/8c2381c39592423db695b64ea1cff438 : SUCCESS in 56m 38s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/81b384cb455b4e8b861e89e6345bac31 : SUCCESS in 52m 48s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/6d9620b433b24d81bf340d47264e45b1 : SUCCESS in 56m 05s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/37c73924fe6c4396ad948e5b42b09579 : SUCCESS in 48m 55s","accounts_in_message":[],"_revision_number":2},{"id":"de87baaa50d23311073a025d6ceea5a632ced8b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 10:13:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d49c9349c1fdd850fe807a4d33821d6f16cc8d75","revisions":{"bc7f32b92b09573cdfb9d44664c1166a085dc8c2":{"kind":"REWORK","_number":1,"created":"2024-07-23 07:54:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/924699/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/99/924699/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/99/924699/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/99/924699/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/99/924699/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/99/924699/1"}}},"commit":{"parents":[{"commit":"c8d7238ae5c332dbef07fe4be7e94f09f00faa1f","subject":"Merge \"Drop unused CONFIG_AMQP_NSS_CERTDB_PW\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c8d7238ae5c332dbef07fe4be7e94f09f00faa1f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 07:38:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 07:54:42.000000000","tz":540},"subject":"Strictly require neutron when nova is enabled","message":"Strictly require neutron when nova is enabled\n\n... and remove a few remaining logics to support deployment without\nneutron.\n\nChange-Id: Id129a3894f3506862ea139c5221ae8bab029320e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bc7f32b92b09573cdfb9d44664c1166a085dc8c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bc7f32b92b09573cdfb9d44664c1166a085dc8c2"}]},"branch":"refs/heads/master"},"d49c9349c1fdd850fe807a4d33821d6f16cc8d75":{"kind":"REWORK","_number":2,"created":"2024-07-23 07:55:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/924699/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/99/924699/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/99/924699/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/99/924699/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/99/924699/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/99/924699/2"}}},"commit":{"parents":[{"commit":"c8d7238ae5c332dbef07fe4be7e94f09f00faa1f","subject":"Merge \"Drop unused CONFIG_AMQP_NSS_CERTDB_PW\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c8d7238ae5c332dbef07fe4be7e94f09f00faa1f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 07:38:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 07:55:42.000000000","tz":540},"subject":"Strictly require neutron when nova is enabled","message":"Strictly require neutron when nova is enabled\n\n... and remove a few remaining logics to support deployment without\nneutron.\n\nChange-Id: Id129a3894f3506862ea139c5221ae8bab029320e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d49c9349c1fdd850fe807a4d33821d6f16cc8d75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/d49c9349c1fdd850fe807a4d33821d6f16cc8d75"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
