)]}'
{"id":"x%2Fpackstack~924743","triplet_id":"x%2Fpackstack~master~I3860f6bd25393ac03b69fffe169e062dc876d55c","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-24 07:38:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3860f6bd25393ac03b69fffe169e062dc876d55c","subject":"Remove logic for mariadb-server \u003c 10.1","status":"MERGED","created":"2024-07-23 15:01:55.000000000","updated":"2024-07-24 07:38:52.000000000","submitted":"2024-07-24 07:38:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924743","meta_rev_id":"e65c03797b729cc5799ebe7c0eb35bcd5fc5ef20","_number":924743,"virtual_id_number":924743,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-24 07:38:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2024-07-24 06:34:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2024-07-24 06:34:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-23 16:14:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-24 06:34:54.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"8ecb60ca7e458ae9dd41820a6c7abf0f653f1fe4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-23 15:01:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6c7f2e07f7bb070f920354711b46410d7951cc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 16:14:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd60a9acfb81488db2dcab8de5f4181b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/906e1895aa9c438e93b47ebdc0101afe : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e7cc5c7340e4b068a8f560e70b4167d : SUCCESS in 3m 20s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/e06ede6f919448d6acb7033fdeba7820 : SUCCESS in 38m 49s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/313b330d43f449028fa724330f9bacce : SUCCESS in 42m 22s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/eaffa20e21ac4af8b3911e709ebf7df9 : SUCCESS in 1h 06m 19s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c4af872c5a9f4c6fbccc66b5d246152e : SUCCESS in 1h 01m 27s","accounts_in_message":[],"_revision_number":1},{"id":"b5ea8787fd03b572f669d8ac4d7df5e5b30c3bab","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-24 06:34:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5c353e1e106612afaa44e50fe43700e2b27aab7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 06:35:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8f19cdd6b83d71d51b227b6c641ddb80a0abd4ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 07:38:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c9410038be947c1b4d37b13a036846a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e6e320d40a343e5a96580fcf69a0ff4 : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ead6240c7fe34be6ad9ab86074431fa8 : SUCCESS in 3m 46s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/0bff00a4da73408ea90a9e5c41232784 : SUCCESS in 51m 39s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/cea3c84eb1964aeaadcbe2cf0e17cc7f : SUCCESS in 48m 49s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/35f49dbbf13c4873938539406ce84186 : SUCCESS in 54m 40s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/7f43c334266f4b288633724802b60d8b : SUCCESS in 57m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e65c03797b729cc5799ebe7c0eb35bcd5fc5ef20","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 07:38:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"10f6ea8fb78868095eb40fe84647b03ee4d5cc1d","revisions":{"10f6ea8fb78868095eb40fe84647b03ee4d5cc1d":{"kind":"REWORK","_number":1,"created":"2024-07-23 15:01:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/924743/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/43/924743/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/43/924743/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/43/924743/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/43/924743/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/43/924743/1"}}},"commit":{"parents":[{"commit":"1533ab12391208aec3ebd4fcf2053c49c89394ea","subject":"Merge \"Strictly require neutron when nova is enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1533ab12391208aec3ebd4fcf2053c49c89394ea"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 15:00:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 15:01:29.000000000","tz":540},"subject":"Remove logic for mariadb-server \u003c 10.1","message":"Remove logic for mariadb-server \u003c 10.1\n\n... because 10.5 is the first version available in CentOS Stream 9.\n\nChange-Id: I3860f6bd25393ac03b69fffe169e062dc876d55c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/10f6ea8fb78868095eb40fe84647b03ee4d5cc1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/10f6ea8fb78868095eb40fe84647b03ee4d5cc1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
