)]}'
{"id":"x%2Fpackstack~924751","triplet_id":"x%2Fpackstack~master~Ia7ecd7e9a0b46b39360adb63291b2a460ca076d0","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-24 07:38:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia7ecd7e9a0b46b39360adb63291b2a460ca076d0","subject":"Drop removed nova options","status":"MERGED","created":"2024-07-23 15:39:22.000000000","updated":"2024-07-24 07:38:54.000000000","submitted":"2024-07-24 07:38:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924751","meta_rev_id":"8d2d24c2f5933df873010e0f6052060ec6d410e8","_number":924751,"virtual_id_number":924751,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-24 07:38:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2024-07-24 06:43:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2024-07-24 06:43:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-23 16:45:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-24 06:43:36.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"67cee7ebc6eea85751bbdc8bbd61a6537123bfee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-23 15:39:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0faa5a4a9a18cce88d7e67aaec97e1c4cdb88ff8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 16:45:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43336d0b799247138e311fa2eb2b7e57\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5045127cdc60439e8a6793676997b348 : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6265018991e54920b09322e264c5fda3 : SUCCESS in 3m 24s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/aa4df924f0c246a7bac60334bb86e054 : SUCCESS in 49m 58s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/8f8026eb16cc4dac9995900beddcdb10 : SUCCESS in 35m 09s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/cea2d35fd2a846b1b469d8e4d310daa4 : SUCCESS in 51m 59s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/d0b30c9211dd4ff1951c680ed5f20b13 : SUCCESS in 1h 00m 09s","accounts_in_message":[],"_revision_number":1},{"id":"3db349cec1985cc217db691f42970a3c634ef4ee","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-24 06:43:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"34b2d62f3ef5c66e81955ebf5e2a6ed8dcecb518","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 06:44:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fc6be81108a3b4b4e44e86154d6e1c75007f124d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 07:38:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ed1a1faed9d4d26bc2d85726198185e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34826db2fc8a4ee2a8e544b735d4004e : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/743a70c709724aa4a280a13e3387cb8b : SUCCESS in 2m 41s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/9414d74114774585846f3b7e69877720 : SUCCESS in 49m 03s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/c4ebe62d82d24346affa871308ddfa15 : SUCCESS in 48m 03s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/1bd5ef8574e34ec6bb513eeb8749c5cd : SUCCESS in 45m 27s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/8e3722bccd464fbd894b6c9592467508 : SUCCESS in 45m 34s","accounts_in_message":[],"_revision_number":1},{"id":"8d2d24c2f5933df873010e0f6052060ec6d410e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 07:38:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff5e71420aa2f1f34aa6de7293e8712dcc9f665e","revisions":{"ff5e71420aa2f1f34aa6de7293e8712dcc9f665e":{"kind":"REWORK","_number":1,"created":"2024-07-23 15:39:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/924751/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/51/924751/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/51/924751/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/51/924751/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/51/924751/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/51/924751/1"}}},"commit":{"parents":[{"commit":"0e96489d101139d4c9331da99af5d689306e5345","subject":"Remove unused netns class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/0e96489d101139d4c9331da99af5d689306e5345"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 15:37:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 15:38:56.000000000","tz":540},"subject":"Drop removed nova options","message":"Drop removed nova options\n\nThe following options were already removed from nova\n - [DEFAULT] metadata_host [1]\n - [DEFAULT] volume_api_class [2]\n\n[1] https://review.opendev.org/c/openstack/nova/+/319530\n[2] https://review.opendev.org/c/openstack/nova/+/696518\n\nChange-Id: Ia7ecd7e9a0b46b39360adb63291b2a460ca076d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ff5e71420aa2f1f34aa6de7293e8712dcc9f665e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ff5e71420aa2f1f34aa6de7293e8712dcc9f665e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
