)]}'
{"id":"x%2Fpackstack~924752","triplet_id":"x%2Fpackstack~master~Ibd2660c977d0c72e117073a686b5b96ba5508b98","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-30 12:58:52.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ibd2660c977d0c72e117073a686b5b96ba5508b98","subject":"Replace non-existing glance_host option","status":"MERGED","created":"2024-07-23 15:44:05.000000000","updated":"2024-07-30 15:23:15.000000000","submitted":"2024-07-30 15:23:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924752","meta_rev_id":"6cc107d4d150c562db65f6d5732788074f88a2fe","_number":924752,"virtual_id_number":924752,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-30 15:23:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2024-07-30 11:31:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2024-07-29 07:39:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2024-07-30 11:31:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-23 16:52:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-29 07:39:58.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-07-30 11:31:07.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"914076eb325feb54f88ed5a025908bf744c88c45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-23 15:44:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f3823434bcc9100b054fbef7e8188a90050ea07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-23 15:47:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"adf595f753f174f0ecb350aa5dc39475fb2443bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-23 15:49:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b9f35e8d49ba50362e48cf9b263154f563ff303b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 16:52:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ef40c0aece241f3a99ffbda2baac5d2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f33caade94f64d3098b60bce854ba752 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/960ac466b4934ababd80393867f46eb4 : SUCCESS in 2m 35s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/94a8f93538944c578d11c8c335cd2139 : SUCCESS in 58m 21s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/7fce85ba12e0464eb7035962eb5bad30 : SUCCESS in 1h 00m 01s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a41debe83ec04249a2d49ef8dfce0790 : SUCCESS in 57m 54s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/bd2361053af54aacbb28e46b8bbd83b7 : SUCCESS in 56m 16s","accounts_in_message":[],"_revision_number":3},{"id":"23495113e6974ddb761294ba4164a46b06d2c1e4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-07-29 07:39:58.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"972854187724c61beb3e0e0ce39a853cdf240b65","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-30 11:31:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7fa382e415f65e241320629edaefa1f60a802fa7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 11:31:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"58b614aa82b97ce6bb9748f88aa4e7a14689de42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 12:43:33.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc90037297434501910ca86c6ea27c69\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d48376534eb7418ea61dccc4af325389 : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb0bdf6df6604a44a7c658392d922b43 : SUCCESS in 3m 17s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/6510f0691d3d4aab99c495c436ae81eb : FAILURE in 1h 05m 21s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/7f02d6a52bd94f52a72dcdfd9d97fd92 : POST_FAILURE in 7m 13s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/f8fbf7949f7f4427b63d0cd3b6b1b9cf : POST_FAILURE in 5m 21s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/59c5e4bcb5384e169ad284ebf26c7550 : POST_FAILURE in 5m 55s","accounts_in_message":[],"_revision_number":3},{"id":"170152a487bb90ff1dfe176ef14daed8eca2152f","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-30 12:50:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"23a2e57009095344698c0e344eed43dc2f29044b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-30 12:58:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8351781bca23731e3a875b41138b7d8d8c540285","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 14:00:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1ee3835ae19415cba8bd5e0803ec337\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c24c6f6c0a124d7384daf494c72076d3 : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe81f775c4874bc58a452d52c6f36526 : SUCCESS in 3m 09s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/49a9b736ed074f198bd86f2afe2e0296 : SUCCESS in 36m 38s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/4465ececcf16497f91268c2e7c99ffe9 : SUCCESS in 56m 37s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/1012440544cf4fdfa99a3eca405b2d20 : SUCCESS in 44m 14s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/374d34a0150c41bbb300846de6769444 : SUCCESS in 48m 57s","accounts_in_message":[],"_revision_number":3},{"id":"8618e9f9228f2fd813bbc93ae6cc055d111f04cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 14:01:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7a9afee52560ed19e1a0a334f8a98d465025ce31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 15:23:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7d3df6565ca44068fc0a269ee1dcc4c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c757c2912fb41a483d92b70000be9ae : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d768f1316dd4dcd9b6aeee71d0127b2 : SUCCESS in 4m 03s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/86f2eb15f796454f929873630257cf18 : SUCCESS in 1h 15m 06s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/a9d6979092f248429c6378269359348a : SUCCESS in 36m 15s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/bd64e064a1d84ffea7cb881b60f67478 : SUCCESS in 40m 36s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/621d2184ad5744598135b30ba32602b9 : SUCCESS in 58m 49s","accounts_in_message":[],"_revision_number":3},{"id":"6cc107d4d150c562db65f6d5732788074f88a2fe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 15:23:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b001514af1665f43c684975d0c5e13a78fe7eb32","revisions":{"e4c2c089e5ff2d5d5eecf91a5d322f324445e929":{"kind":"REWORK","_number":1,"created":"2024-07-23 15:44:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/924752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/52/924752/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/52/924752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/52/924752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/52/924752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/52/924752/1"}}},"commit":{"parents":[{"commit":"1533ab12391208aec3ebd4fcf2053c49c89394ea","subject":"Merge \"Strictly require neutron when nova is enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1533ab12391208aec3ebd4fcf2053c49c89394ea"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 15:41:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 15:42:07.000000000","tz":540},"subject":"Replace non-existing glance_host option","message":"Replace non-existing glance_host option\n\nThe [DEFAULT] glance_host option was removed from cinder in favor of\nthe detection through catalog[1]. Also the option has never been\nimplemented in manila.\n\n[1] https://review.opendev.org/c/openstack/cinder/+/252476\n\nChange-Id: Ibd2660c977d0c72e117073a686b5b96ba5508b98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e4c2c089e5ff2d5d5eecf91a5d322f324445e929"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/e4c2c089e5ff2d5d5eecf91a5d322f324445e929"}]},"branch":"refs/heads/master"},"8083ea6f7d59468930a9f24b724d00ea2cfa6660":{"kind":"REWORK","_number":2,"created":"2024-07-23 15:47:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/924752/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/52/924752/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/52/924752/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/52/924752/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/52/924752/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/52/924752/2"}}},"commit":{"parents":[{"commit":"1533ab12391208aec3ebd4fcf2053c49c89394ea","subject":"Merge \"Strictly require neutron when nova is enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1533ab12391208aec3ebd4fcf2053c49c89394ea"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 15:41:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 15:47:17.000000000","tz":540},"subject":"Replace non-existing glance_host option","message":"Replace non-existing glance_host option\n\nThe [DEFAULT] glance_host option was removed from cinder and ironic in\nfavor of the detection through catalog[1][2]. Also the option has never\nbeen implemented in manila.\n\n[1] https://review.opendev.org/c/openstack/cinder/+/252476\n[2] https://review.opendev.org/c/openstack/ironic/+/476498\n\nChange-Id: Ibd2660c977d0c72e117073a686b5b96ba5508b98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8083ea6f7d59468930a9f24b724d00ea2cfa6660"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/8083ea6f7d59468930a9f24b724d00ea2cfa6660"}]},"branch":"refs/heads/master"},"b001514af1665f43c684975d0c5e13a78fe7eb32":{"kind":"REWORK","_number":3,"created":"2024-07-23 15:49:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/924752/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/52/924752/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/52/924752/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/52/924752/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/52/924752/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/52/924752/3"}}},"commit":{"parents":[{"commit":"1533ab12391208aec3ebd4fcf2053c49c89394ea","subject":"Merge \"Strictly require neutron when nova is enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1533ab12391208aec3ebd4fcf2053c49c89394ea"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 15:41:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 15:48:52.000000000","tz":540},"subject":"Replace non-existing glance_host option","message":"Replace non-existing glance_host option\n\nThe [DEFAULT] glance_host option was removed from cinder and ironic in\nfavor of the detection through catalog[1][2]. Also the option has never\nbeen implemented in manila.\n\n[1] https://review.opendev.org/c/openstack/cinder/+/252476\n[2] https://review.opendev.org/c/openstack/ironic/+/476498\n\nChange-Id: Ibd2660c977d0c72e117073a686b5b96ba5508b98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b001514af1665f43c684975d0c5e13a78fe7eb32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/b001514af1665f43c684975d0c5e13a78fe7eb32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
