)]}'
{"id":"x%2Fpackstack~924753","triplet_id":"x%2Fpackstack~master~I5d2404e16cce69f1897e3b3e5714e43ab24b7ed1","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-24 12:49:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5d2404e16cce69f1897e3b3e5714e43ab24b7ed1","subject":"Replace _member_ role by member role","status":"MERGED","created":"2024-07-23 15:55:56.000000000","updated":"2024-07-24 12:49:15.000000000","submitted":"2024-07-24 12:49:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924753","meta_rev_id":"41235738d7335dbed92ae39053a3ce05e7eca775","_number":924753,"virtual_id_number":924753,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-24 12:49:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2024-07-24 11:41:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2024-07-24 11:41:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-23 17:15:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-24 11:41:12.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"9ff0de298af8a6064d14e477b1be66a825e45c33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-23 15:55:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa2c94fe4d44406117781303321531f570cdfbc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-23 16:02:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c88f7b8fbe5100a2e139eb0563785160b13b46d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 17:15:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81e3a91d3a5549a494294dbfdc5bab83\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bee30b6c37cb41ab85d1e53f9c5df633 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebb8ef4928ae4dc7a4e4cf4c7a1b050c : SUCCESS in 3m 20s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/9e07f2d2c2334124a33d57f623c515ce : SUCCESS in 37m 10s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/dbf3e0121b1a4bc3aaf49d6af01027b0 : SUCCESS in 38m 13s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a182985c93f0434c8d669d7a2ba2fd12 : SUCCESS in 41m 13s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/b820dfc363f848f7b42dc62143d9b643 : SUCCESS in 1h 06m 02s","accounts_in_message":[],"_revision_number":2},{"id":"9e5b1829f6ca68247f3e55a0d241e39eec4672e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-24 10:21:56.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"fca416c229c0dab7c2dd83b87ddae8b3373b92b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 11:23:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86b97b74d3bf44eaa07362bcec3f6f2c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16b08a484398435ebbe8d43448496b5e : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3d4c43531ed491c8cb259c3c03abdc0 : SUCCESS in 3m 06s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/4bc9b8c7f8ea42498bc7ca7c57805ece : SUCCESS in 46m 40s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/1a57b68510e149aa88aa303724efb2a8 : SUCCESS in 55m 06s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/802828a11b104b688dea938bf2b3c9bf : SUCCESS in 51m 23s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/49c1089b09614b83a969240b03e8315c : SUCCESS in 46m 21s","accounts_in_message":[],"_revision_number":3},{"id":"99ea5e10c42291517e9eb4f49757788ae14c1992","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-24 11:41:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"152ca17fe66e3b5c4ff4740f22623e7006e11e3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 11:42:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ed15a7912c928663b80d09eb0535da14ededdcb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 12:49:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/042909331b9f4c18843d782cfe42c1ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da07f940f6a640e489e3ab9ef6ce77c5 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70ae65fa49f3449082cbde0f07e0859d : SUCCESS in 3m 31s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/4a61adb8f2db4e419a153d2477ddadc6 : SUCCESS in 54m 18s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/154e3d863db54af49b228a0a651e3590 : SUCCESS in 50m 37s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/95ccb22bbede4f4c8722df5d9762b56a : SUCCESS in 58m 54s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/cf719f9a0f154dc5967283fb3440cc43 : SUCCESS in 58m 25s","accounts_in_message":[],"_revision_number":3},{"id":"41235738d7335dbed92ae39053a3ce05e7eca775","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 12:49:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1804ec25cb212be53c48780ef0385ea0e91522b9","revisions":{"19d33cde630d4aab52a82f97a3afa5eb666586d0":{"kind":"REWORK","_number":1,"created":"2024-07-23 15:55:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/924753/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/53/924753/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/53/924753/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/53/924753/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/53/924753/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/53/924753/1"}}},"commit":{"parents":[{"commit":"1533ab12391208aec3ebd4fcf2053c49c89394ea","subject":"Merge \"Strictly require neutron when nova is enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1533ab12391208aec3ebd4fcf2053c49c89394ea"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 15:52:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 15:55:10.000000000","tz":540},"subject":"Replace _member_ role by member role","message":"Replace _member_ role by member role\n\nThe member role is now commonly used as one of the default roles\ncreated during bootstrap. Replace the legacy _member_ role by it.\n\nChange-Id: I5d2404e16cce69f1897e3b3e5714e43ab24b7ed1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/19d33cde630d4aab52a82f97a3afa5eb666586d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/19d33cde630d4aab52a82f97a3afa5eb666586d0"}]},"branch":"refs/heads/master"},"4dc3f8acb49591d010ec4489545f7ce6b7993f7c":{"kind":"REWORK","_number":2,"created":"2024-07-23 16:02:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/924753/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/53/924753/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/53/924753/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/53/924753/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/53/924753/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/53/924753/2"}}},"commit":{"parents":[{"commit":"a89355ba5f1b34e295f403d65c517a01af518eca","subject":"Remove trusts_delegated_roles role from the user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a89355ba5f1b34e295f403d65c517a01af518eca"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 15:52:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 16:02:26.000000000","tz":540},"subject":"Replace _member_ role by member role","message":"Replace _member_ role by member role\n\nThe member role is now commonly used as one of the default roles\ncreated during bootstrap. Replace the legacy _member_ role by it.\n\nChange-Id: I5d2404e16cce69f1897e3b3e5714e43ab24b7ed1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4dc3f8acb49591d010ec4489545f7ce6b7993f7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/4dc3f8acb49591d010ec4489545f7ce6b7993f7c"}]},"branch":"refs/heads/master"},"1804ec25cb212be53c48780ef0385ea0e91522b9":{"kind":"REWORK","_number":3,"created":"2024-07-24 10:21:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/924753/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/53/924753/3","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/53/924753/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/53/924753/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/53/924753/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/53/924753/3"}}},"commit":{"parents":[{"commit":"f9ff68b48d9ead128bc6562e53ffec54dbfccb5f","subject":"Remove trusts_delegated_roles role from the user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f9ff68b48d9ead128bc6562e53ffec54dbfccb5f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 15:52:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-24 10:21:45.000000000","tz":540},"subject":"Replace _member_ role by member role","message":"Replace _member_ role by member role\n\nThe member role is now commonly used as one of the default roles\ncreated during bootstrap. Replace the legacy _member_ role by it.\n\nChange-Id: I5d2404e16cce69f1897e3b3e5714e43ab24b7ed1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1804ec25cb212be53c48780ef0385ea0e91522b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1804ec25cb212be53c48780ef0385ea0e91522b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
