)]}'
{"id":"x%2Fpackstack~924757","triplet_id":"x%2Fpackstack~master~I3331ccb76a0281f747a36332341cc3e62e276ed1","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2024-07-24 12:43:57.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-24 10:45:04.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I3331ccb76a0281f747a36332341cc3e62e276ed1","subject":"Remove trusts_delegated_roles role from the user","status":"MERGED","created":"2024-07-23 16:02:37.000000000","updated":"2024-07-24 12:43:57.000000000","submitted":"2024-07-24 12:43:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924757","meta_rev_id":"3b98dc99df807e232859afa48813f7cfb033a4c4","_number":924757,"virtual_id_number":924757,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-24 12:43:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2024-07-24 10:21:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2024-07-24 10:45:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-23 17:00:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-24 10:12:44.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"2afb4765a54166b618deb476f0b255450705ed4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-23 16:02:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55b710798cb0dd85296a8951078bd7cba40389dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 17:00:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02383d1e66d0474b98c3562aa0436a7b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bce8da9b8888482da9c0ceee4701aee5 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b40bfd8e9e34a7f970eceeb69981edd : SUCCESS in 3m 15s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/5edab5ea8bee41aaaa2b16d7e8aaf7be : SUCCESS in 52m 04s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/eef3ca7c3e214771bfc179b48f2a455d : SUCCESS in 49m 30s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/4d64602e8fc14c71b4e0eb99987c8c11 : SUCCESS in 37m 02s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/1bb2cd18db61411e9e104b87997a5910 : SUCCESS in 46m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e2ef9e7477dd6a3362f61a5cf48c3f10a60cd020","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-24 10:12:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a2cb562a5719d24950aafbe2f89c3d6ef5b199e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 10:13:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"82fda24d71cfdc496d6cbaf434aa25006a8f2577","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-24 10:21:56.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a7a8d8fab0eb28bc178ad48a347bb463277173a9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-24 10:45:04.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"799ba7aa4c2c75ae698949639e48a33d14830fc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 11:41:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96e854b4dad449618c0d853d6ee00b00\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41cb848202bc48b49614565bfc326d07 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bf67c97e2864ae2968b43af5ad6925a : SUCCESS in 3m 08s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/ce4b5af5da9246419928e917ac04148d : SUCCESS in 39m 16s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/95bee6116a694cbd910b9a0ada47ec6a : SUCCESS in 53m 31s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/656ccf11abc4443dbfe9187138e86f0b : SUCCESS in 1h 09m 24s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/801b3691ecdd4d25831ab26b0234e752 : SUCCESS in 58m 14s","accounts_in_message":[],"_revision_number":2},{"id":"fbdef4200e1f525f58117b146290c1099abb3e63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 11:42:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"551b770aa284627cd8110dbf7cb138a2b348e951","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 12:43:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebdc01380f8d4af48888c6fee6083c81\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52ab1eaee281472a84f1b5f6504beace : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7109522aa0174202996bb666778ea55b : SUCCESS in 3m 00s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/e8b62e0134464d2e808ff2b361f87135 : SUCCESS in 40m 32s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/92cb1426aac3470495f1363eb2bf7204 : SUCCESS in 52m 01s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/8f36cec808c5460f90848f0024ca002a : SUCCESS in 41m 07s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/7b8af064497843dab58e727e228e5968 : SUCCESS in 56m 02s","accounts_in_message":[],"_revision_number":2},{"id":"3b98dc99df807e232859afa48813f7cfb033a4c4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 12:43:57.000000000","message":"Change has been successfully merged\n\n1 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f9ff68b48d9ead128bc6562e53ffec54dbfccb5f","revisions":{"a89355ba5f1b34e295f403d65c517a01af518eca":{"kind":"REWORK","_number":1,"created":"2024-07-23 16:02:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/924757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/57/924757/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/57/924757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/57/924757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/57/924757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/57/924757/1"}}},"commit":{"parents":[{"commit":"1533ab12391208aec3ebd4fcf2053c49c89394ea","subject":"Merge \"Strictly require neutron when nova is enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1533ab12391208aec3ebd4fcf2053c49c89394ea"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 15:59:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 16:01:59.000000000","tz":540},"subject":"Remove trusts_delegated_roles role from the user","message":"Remove trusts_delegated_roles role from the user\n\nThis role is not created/used by default.\n\nChange-Id: I3331ccb76a0281f747a36332341cc3e62e276ed1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a89355ba5f1b34e295f403d65c517a01af518eca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/a89355ba5f1b34e295f403d65c517a01af518eca"}]},"branch":"refs/heads/master"},"f9ff68b48d9ead128bc6562e53ffec54dbfccb5f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-24 10:21:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/924757/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/57/924757/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/57/924757/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/57/924757/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/57/924757/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/57/924757/2"}}},"commit":{"parents":[{"commit":"759b68c8d27a4e86ba7a0bb24bd7e3469c139084","subject":"Switch to structure facts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/759b68c8d27a4e86ba7a0bb24bd7e3469c139084"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 15:59:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-24 10:21:27.000000000","tz":540},"subject":"Remove trusts_delegated_roles role from the user","message":"Remove trusts_delegated_roles role from the user\n\nThis role is not created/used by default.\n\nChange-Id: I3331ccb76a0281f747a36332341cc3e62e276ed1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f9ff68b48d9ead128bc6562e53ffec54dbfccb5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/f9ff68b48d9ead128bc6562e53ffec54dbfccb5f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
