)]}'
{"id":"x%2Fpackstack~924842","triplet_id":"x%2Fpackstack~master~I0bd64aa79023e65b8d81851c59e7b3ff8ccfb56a","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-24 14:13:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0bd64aa79023e65b8d81851c59e7b3ff8ccfb56a","subject":"Horizon: Switch to PyMemcacheCache backend","status":"MERGED","created":"2024-07-24 10:29:21.000000000","updated":"2024-07-24 14:13:58.000000000","submitted":"2024-07-24 14:13:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924842","meta_rev_id":"fcefc41e2e153c073c06f58e0ed2144d2d517cd5","_number":924842,"virtual_id_number":924842,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-24 14:13:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2024-07-24 13:04:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2024-07-24 13:04:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-24 11:26:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-24 13:04:36.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"a4fc6e546324b3ec42f461919264f83903f91796","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-24 10:29:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89534cd0c8dbd4eeaa2f7748fea465ef9f465da5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 11:26:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b1feffe805e437790839f4d3121b948\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88277dd353ba4f14a516cdafa636edf1 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74a3acb3fea643638ee6e099c4a779a7 : SUCCESS in 2m 37s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/1de7f206fd59484793ac4d8c0e1e63c3 : SUCCESS in 38m 39s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/3362b40971a04420b7a0f524fca302bd : SUCCESS in 39m 55s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/7ec8a68d46d1434db15e0a55b9919879 : SUCCESS in 40m 33s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/cf29728d2b6c4963a358e5c6e3837c26 : SUCCESS in 39m 35s","accounts_in_message":[],"_revision_number":1},{"id":"35f03a8cb7437e70abf37f9e8d2982e2a49b3a6e","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-24 13:04:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d29126e1c9b51134260845f802b09d3b5b76525b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 13:05:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b6ef440df6d5d348029d2bc8ef40a7c3e541452d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 14:13:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50a79570fdfc4348bf161089b6517504\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7d6939b011a48129cb62ba107ce03df : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2463e7be526149ce941c738f06f9848d : SUCCESS in 3m 56s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/88c9b7bdbaff4acf992e5fb633cd4e1d : SUCCESS in 54m 25s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/0dbd722cec8742ecbee7d40d63c18e67 : SUCCESS in 55m 23s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/a6db46c22c7c49a99b29cd80f69c90d6 : SUCCESS in 57m 16s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/2e4509b402394302bf253b25dc7ad1cc : SUCCESS in 1h 04m 00s","accounts_in_message":[],"_revision_number":1},{"id":"fcefc41e2e153c073c06f58e0ed2144d2d517cd5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 14:13:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e36bee28f2d38f0f88277df4fd9546964f3a5f4","revisions":{"1e36bee28f2d38f0f88277df4fd9546964f3a5f4":{"kind":"REWORK","_number":1,"created":"2024-07-24 10:29:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/924842/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/42/924842/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/42/924842/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/42/924842/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/42/924842/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/42/924842/1"}}},"commit":{"parents":[{"commit":"759b68c8d27a4e86ba7a0bb24bd7e3469c139084","subject":"Switch to structure facts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/759b68c8d27a4e86ba7a0bb24bd7e3469c139084"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-24 10:28:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-24 10:29:10.000000000","tz":540},"subject":"Horizon: Switch to PyMemcacheCache backend","message":"Horizon: Switch to PyMemcacheCache backend\n\nThe MemcachedCache backend was deprecated in django 3.2 and was removed\nin 4.1.\n\nChange-Id: I0bd64aa79023e65b8d81851c59e7b3ff8ccfb56a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1e36bee28f2d38f0f88277df4fd9546964f3a5f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/1e36bee28f2d38f0f88277df4fd9546964f3a5f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
