)]}'
{"id":"x%2Fpackstack~924843","triplet_id":"x%2Fpackstack~master~I10bcad01ba1add61599e7f4759242c36b97dc927","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-24 16:01:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I10bcad01ba1add61599e7f4759242c36b97dc927","subject":"Drop remaining usage of CONFIG_HORIZON_NEUTRON_LB","status":"MERGED","created":"2024-07-24 10:32:55.000000000","updated":"2024-07-24 16:01:49.000000000","submitted":"2024-07-24 16:01:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924843","meta_rev_id":"10043e7146b526d0c7dbcc34ba6709a954a624c2","_number":924843,"virtual_id_number":924843,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-24 16:01:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2024-07-24 14:51:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2024-07-24 14:51:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-24 11:30:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-24 14:51:03.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"46057f414d519f250a64c9705a2d9e802106c5f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-24 10:32:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60cd3f0957f97641b85354d145ad4c7a451b8efe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 11:30:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e081d546d1246c88415c46d3183b330\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbcb7da583684d11941cb2f7a996f324 : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4021c856dc84139bec757a06ed213f3 : SUCCESS in 3m 00s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/adfca14bc7d0442781f860a6fcaf7cf4 : SUCCESS in 38m 01s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/70899680c52d471cbc09d916ca79d5d7 : SUCCESS in 47m 32s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/93006ec805d3480db84de92e803d7308 : SUCCESS in 56m 16s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/7787a2f3c7234c5abea641694678ce35 : SUCCESS in 42m 31s","accounts_in_message":[],"_revision_number":1},{"id":"b43ae9e9f16d7728f30e8e2695d95b047a5e796e","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-24 14:51:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"26bd22a30cf6d37f1803be7dce9b9be40d8b95cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 14:51:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9fde64407900e3fb175f27e55633b0528dac7e28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 16:01:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6793f1bf7616401babad527b097be238\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8ed6a950a0044b78d9219d3ddec3747 : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de839f76dddb472f9a5e802dbeac91a7 : SUCCESS in 3m 43s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/a7275288401241a4af60137510309439 : SUCCESS in 56m 21s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/9461c8efb418460bbdb05b468c35d9c5 : SUCCESS in 36m 20s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/b7a4b567930c4a2999fb2bf9474960a0 : SUCCESS in 1h 00m 52s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/0dcd3b59b5a6460ab20fdcc01c510aca : SUCCESS in 59m 21s","accounts_in_message":[],"_revision_number":1},{"id":"10043e7146b526d0c7dbcc34ba6709a954a624c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 16:01:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5459f795716aeb8393315a773623e8b6d7568d32","revisions":{"5459f795716aeb8393315a773623e8b6d7568d32":{"kind":"REWORK","_number":1,"created":"2024-07-24 10:32:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/924843/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/43/924843/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/43/924843/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/43/924843/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/43/924843/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/43/924843/1"}}},"commit":{"parents":[{"commit":"759b68c8d27a4e86ba7a0bb24bd7e3469c139084","subject":"Switch to structure facts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/759b68c8d27a4e86ba7a0bb24bd7e3469c139084"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-24 10:32:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-24 10:32:27.000000000","tz":540},"subject":"Drop remaining usage of CONFIG_HORIZON_NEUTRON_LB","message":"Drop remaining usage of CONFIG_HORIZON_NEUTRON_LB\n\nThis option is no longer used since Neutron LBaaS was retired.\n\nChange-Id: I10bcad01ba1add61599e7f4759242c36b97dc927\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5459f795716aeb8393315a773623e8b6d7568d32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5459f795716aeb8393315a773623e8b6d7568d32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
