)]}'
{"id":"x%2Fpackstack~924900","triplet_id":"x%2Fpackstack~master~I4a8b476f0573f5fb771174ab86fad293462df060","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-30 12:58:44.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I4a8b476f0573f5fb771174ab86fad293462df060","subject":"Redis: Properly disable unix socket","status":"MERGED","created":"2024-07-25 04:08:33.000000000","updated":"2024-07-30 15:24:10.000000000","submitted":"2024-07-30 15:24:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924900","meta_rev_id":"ad4765c22933896d66197d60632b6fce2b2116b4","_number":924900,"virtual_id_number":924900,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-30 15:24:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":2,"date":"2024-07-30 11:22:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2024-07-29 07:39:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":1,"date":"2024-07-30 11:22:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-25 05:18:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-29 07:39:23.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-07-30 11:22:50.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"456fc08d8d78c046423d46f8ea0f4507d7e06ec3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-25 04:08:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"506320259b2200c577863f56bd32f5b7429a1d49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 05:18:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efe0475f7005429f98e99a0345e053b4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90047a40d13a4070bfffb29439663eb0 : SUCCESS in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efd559754962427aae7a24c0eeed873e : SUCCESS in 4m 04s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/0a0669e06d96466d87c3b09a8eeb8d17 : SUCCESS in 31m 19s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/b7b77d9159ed4db0b49a30536bbe723f : SUCCESS in 54m 46s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/7285a86375d24f7eaf84ea827656fbda : SUCCESS in 1h 04m 22s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/ee1128354e2c422b970a0ad8ebc21968 : SUCCESS in 1h 04m 29s","accounts_in_message":[],"_revision_number":1},{"id":"18b290dd6b917d9818dcda84d7f8f0008ac8b7a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-25 07:08:35.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d579f68a12496c47d6dc5a292a3e2b250b6540a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 08:09:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfc9eb05e1cd43bc99c09b1cf1bf5945\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39d3499e493446dbae1c2c8401c679e9 : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0464fe87fb24ec5875a9fd4267cc1bb : SUCCESS in 2m 55s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/7d4dbffe8d014460899673dbf56b0703 : SUCCESS in 55m 02s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/8ee7337437714c9faec4620a787296ca : SUCCESS in 52m 55s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/d0783dd9290b48b2acdde66fdb5b4713 : SUCCESS in 55m 29s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/26674f9aaad34bba920c9b360e34487f : SUCCESS in 55m 39s","accounts_in_message":[],"_revision_number":2},{"id":"5a31464b277db9744f3b0ac359be49a319d31313","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-07-29 07:39:23.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"24ae5076a792ff64b5b8b7e176b0fde1dbdcdeb8","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-30 11:22:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"45701914a14151e5e7398a81d28b96090379a5f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 11:23:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a1cd7dd47272d9e8b693b4f7063c6c25814db72a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 11:45:57.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/924906 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"cdda2cb19c54dc6e1217c1f4307409740a39dab1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-30 12:58:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8cdccb0f999c9a95bb1a4ba43b77b912968a2da4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 13:49:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e12b94977d614377b3ed6525ecd9c692\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29b58cc87506460d868ad481b9187875 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db9f84c01cb54d84bb24462058ac169b : SUCCESS in 4m 09s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/91f5cd95542145e5a44de644d673a316 : SUCCESS in 39m 40s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/9cafcff8e24543eb8aadaeca8801c1b6 : SUCCESS in 42m 22s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/3236a08f5f6e441b9aa738a485c4d498 : SUCCESS in 44m 19s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/318eeb1f49924a239a5cef80162dfef0 : SUCCESS in 47m 54s","accounts_in_message":[],"_revision_number":2},{"id":"a09ac2e77e62e20a3629236f7542fdcf5f8e5ee7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 14:20:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1bba26b15c74b8a82752a222d0390b68d24075e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 15:24:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c04c14e5ebd74f079551674cccaf42dc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19db729e0fb64d18b8fc4e3010b752a3 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d16539e6d76a44e59795c19b5b080374 : SUCCESS in 3m 31s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/bcfa977acd554fe3a72f5971a985a00d : SUCCESS in 47m 06s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/3f2774c0b2a34c4991abbb604217f8dc : SUCCESS in 35m 53s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/381c1afa5f714a869a01a40d6591d734 : SUCCESS in 40m 07s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/47890fab901744b28a1d60cadfa02df0 : SUCCESS in 55m 06s","accounts_in_message":[],"_revision_number":2},{"id":"ad4765c22933896d66197d60632b6fce2b2116b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 15:24:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"39c6ad46b05392188b3759ba3e2021a4052d829b","revisions":{"60a398d822798d2328780d1bb7db1e172c198a45":{"kind":"REWORK","_number":1,"created":"2024-07-25 04:08:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/924900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/00/924900/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/00/924900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/00/924900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/00/924900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/00/924900/1"}}},"commit":{"parents":[{"commit":"5f075b3b0562b856f5e420f30cec5fa437daf6eb","subject":"Merge \"Drop remaining usage of CONFIG_HORIZON_NEUTRON_LB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/5f075b3b0562b856f5e420f30cec5fa437daf6eb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-25 04:04:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-25 04:08:23.000000000","tz":540},"subject":"Redis: Properly disable unix socket","message":"Redis: Properly disable unix socket\n\nundef is silently translated to the default value by Puppet, thus\nthe socket was not disabled as intended.\n\nRecent puppet-redis accpets \u0027\u0027 as  the indicator to disable unix domain\nsocket so we can use it instead. Also drop the _perm option which is\nnot actually used when unix domain socket is disabled.\n\nChange-Id: I4a8b476f0573f5fb771174ab86fad293462df060\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/60a398d822798d2328780d1bb7db1e172c198a45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/60a398d822798d2328780d1bb7db1e172c198a45"}]},"branch":"refs/heads/master"},"39c6ad46b05392188b3759ba3e2021a4052d829b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-25 07:08:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/924900/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/00/924900/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/00/924900/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/00/924900/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/00/924900/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/00/924900/2"}}},"commit":{"parents":[{"commit":"c0aac57ecadf89ee5cb96aaac87b9133c86967ed","subject":"Capture redis config files and logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/c0aac57ecadf89ee5cb96aaac87b9133c86967ed"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-25 04:04:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-25 07:08:26.000000000","tz":540},"subject":"Redis: Properly disable unix socket","message":"Redis: Properly disable unix socket\n\nundef is silently translated to the default value by Puppet, thus\nthe socket was not disabled as intended.\n\nRecent puppet-redis accpets \u0027\u0027 as  the indicator to disable unix domain\nsocket so we can use it instead. Also drop the _perm option which is\nnot actually used when unix domain socket is disabled.\n\nChange-Id: I4a8b476f0573f5fb771174ab86fad293462df060\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/39c6ad46b05392188b3759ba3e2021a4052d829b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/39c6ad46b05392188b3759ba3e2021a4052d829b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
