)]}'
{"id":"x%2Fpackstack~925192","triplet_id":"x%2Fpackstack~stable%2F2024.1~I681164fd22a4f72c4cb4ff2d7ad3428b2c715002","project":"x/packstack","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2024-07-31 14:30:23.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-31 13:23:10.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2024-07-31 14:30:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I681164fd22a4f72c4cb4ff2d7ad3428b2c715002","subject":"Switch to stable/2024.1 branh","status":"MERGED","created":"2024-07-30 12:03:32.000000000","updated":"2024-07-31 14:30:23.000000000","submitted":"2024-07-31 14:30:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925192","meta_rev_id":"50747bf4ee165f0e6a7b99a3ea675a08b3e64791","_number":925192,"virtual_id_number":925192,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-31 14:30:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-07-31 13:23:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-07-31 13:23:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-30 13:08:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-31 08:55:36.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-07-31 08:55:43.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"}],"messages":[{"id":"477a7e9ca9e4d6dc60b8ed1939cc379a8fc3efc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-30 12:03:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c05b42db3b0fb4325e3d2cb45e530ba59d0100b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 13:08:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71ee16415f5d4a00b27ea6f77163cb60\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee2eb5b0ab0d46dcb2ea7ac61bfb381b : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1a99f0be6244cdb9dc6175126dd2bd3 : SUCCESS in 4m 13s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/c8a1cb0401c24ff4820431d5ab990652 : SUCCESS in 54m 29s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/daa996eb07dc40a284e0e8e0e2be9c28 : SUCCESS in 55m 24s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/38caa849349b4e3f995c1fde8654404a : SUCCESS in 54m 19s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/e199f381ef8d4a94ad67c233a84ad8f2 : POST_FAILURE in 6m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f29bee2db22e5eacefd16a0c8cd23921f197389b","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-30 15:27:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"40ddb9f34229ebbde61a08326a3f9ff448836f2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 16:31:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0742f3fcc08846db9921d5586c6b26bf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/439cfe0999564baaabdb4d554a6f53f9 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6313673c64043158d7ea93a4c45b98f : SUCCESS in 3m 03s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/8cda941935c84af5a6cf9b27f4951cdd : SUCCESS in 43m 35s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/48f2fc3d456c42c69437867ef40c97f6 : SUCCESS in 52m 48s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/b8d02e973ed7488abda3a2af4a2d5014 : SUCCESS in 54m 21s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/c5b4618f34194937b9d4da9787ba5a7b : SUCCESS in 53m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ab8e5599db0907c264909b6642ab8695c00eced0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-31 13:23:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"73b2cc9bec3dad8a7415f8af8670c2f9b0027395","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 13:23:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c4b951e3912e7aaa492f72cc51c724dcc1b85db0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 14:30:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a68ad239aa8b4670a07aac9d984de066\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d377e1b222d4751814e949ce6dc349e : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01c8108aa73e41268fb1312256a13edd : SUCCESS in 2m 46s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/a41938f6e4454b4babe9d6d99108521f : SUCCESS in 40m 56s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/2a9836d944884dcba87ef91f19fd0f2f : SUCCESS in 33m 00s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/c03828f6241247c8a59b75bcc2ea0c3e : SUCCESS in 1h 01m 16s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/546fd3f31cf84986ae05fbfd66abb2f9 : SUCCESS in 55m 33s","accounts_in_message":[],"_revision_number":1},{"id":"50747bf4ee165f0e6a7b99a3ea675a08b3e64791","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 14:30:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6d0c28e72298979427c4156bd51e96edc5e7e0eb","revisions":{"6d0c28e72298979427c4156bd51e96edc5e7e0eb":{"kind":"REWORK","_number":1,"created":"2024-07-30 12:03:32.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/92/925192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/92/925192/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/92/925192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/92/925192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/92/925192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/92/925192/1"}}},"commit":{"parents":[{"commit":"ca8d297ff3ab8c1aee8fea05f80fc3c5df54cb7d","subject":"Merge \"Nova: Enable service token\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/ca8d297ff3ab8c1aee8fea05f80fc3c5df54cb7d"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-07-30 12:02:48.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-07-30 12:03:20.000000000","tz":120},"subject":"Switch to stable/2024.1 branh","message":"Switch to stable/2024.1 branh\n\nChange-Id: I681164fd22a4f72c4cb4ff2d7ad3428b2c715002\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6d0c28e72298979427c4156bd51e96edc5e7e0eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6d0c28e72298979427c4156bd51e96edc5e7e0eb"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
