)]}'
{"id":"x%2Fpackstack~930807","triplet_id":"x%2Fpackstack~master~Ie100fa47472c2bd55635d5b4a524cf9ef6671940","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-03 14:13:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie100fa47472c2bd55635d5b4a524cf9ef6671940","subject":"Fix undefined names","status":"MERGED","created":"2024-09-28 12:13:15.000000000","updated":"2024-10-03 14:13:38.000000000","submitted":"2024-10-03 14:13:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930807","meta_rev_id":"0deea0330958567a34ca7269d5189e752bc5ce2c","_number":930807,"virtual_id_number":930807,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-03 14:13:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"recommended":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"all":[{"value":1,"date":"2024-09-30 07:46:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-10-01 21:59:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":2,"date":"2024-10-03 13:13:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":1,"date":"2024-10-03 13:13:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-28 13:47:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-30 07:46:56.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-10-01 21:59:10.000000000","updated_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"reviewer":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"state":"REVIEWER"},{"updated":"2024-10-03 13:13:12.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"}],"messages":[{"id":"4a80833d805a60a302fc8d114041cf4fcd22e7ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-28 12:13:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8bce4b094e9bc62d8a9b6086d51eae75576ad34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-28 12:43:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4c2e20da0ee6d0cd1cd83466a4b328c0bdb3f6fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-28 13:47:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3ccd7d93344491e85c2766cf766676c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c18394c316be490f9266c46afd2923f0 : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4686115de30d4a3a8eae2aa14d547c92 : SUCCESS in 2m 56s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/585d585ce73e4bc5aecb10d9c53db0f9 : SUCCESS in 55m 19s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/c1f28d16c7234ae8a71b4f2162290a3b : SUCCESS in 48m 02s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/eb9d3c5069bf40e0b7505f44b62f6364 : SUCCESS in 59m 09s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/92050a30c1184dfa8c65f8195b52a204 : SUCCESS in 59m 35s","accounts_in_message":[],"_revision_number":2},{"id":"af2c561d7c83aea15aec177fd95395911229963f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-09-30 07:46:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"90b6811b12a3a7eaa61be663b03873028a8f21fe","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2024-10-01 21:59:10.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"063ae308b49fbec66f5f15e053399b47253a40e3","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-10-03 13:13:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9489151755d697562c9e1dbb9f3fef388d7ea2ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 13:13:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9d3768f5438387de92c5eea20d0394a4be131ef0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 14:13:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ad87cba9d6b462582d2a145e831b1a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bbe1b8c9f71460c82fb98be67957557 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d1126d2c727427c9dec5ddef7ef7506 : SUCCESS in 3m 31s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/4fe93a566a6244448aad5ed524b1576d : SUCCESS in 51m 00s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/5ecf03fe22e9445bb379946ec88fe155 : SUCCESS in 44m 34s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/0ea3b2f70b284a70acb35d673525a0c8 : SUCCESS in 55m 16s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/cb51b306c2624cfd915b944a439aa85f : SUCCESS in 41m 06s","accounts_in_message":[],"_revision_number":2},{"id":"0deea0330958567a34ca7269d5189e752bc5ce2c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 14:13:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bd593b8df76db090a7bafc1032175475c90a25a3","revisions":{"aaef82ada428f299a6c529917bea905fce1edaa8":{"kind":"REWORK","_number":1,"created":"2024-09-28 12:13:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/930807/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/07/930807/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/07/930807/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/07/930807/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/07/930807/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/07/930807/1"}}},"commit":{"parents":[{"commit":"6253bb59ce168b841900049ab68531e89c1993eb","subject":"Merge \"Bump hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6253bb59ce168b841900049ab68531e89c1993eb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-28 12:11:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-28 12:13:07.000000000","tz":540},"subject":"Fix undefined names","message":"Fix undefined names\n\nThis fixes apparent problems detected when F821 check is enabled.\n\n./packstack/installer/core/drones.py:43:16: F821 undefined name \u0027ScriptRuntimeError\u0027\n        except ScriptRuntimeError as ex:\n               ^\n./packstack/installer/core/drones.py:313:9: F821 undefined name \u0027shutil\u0027\n        shutil.rmtree(self.local_tmpdir, ignore_errors\u003dTrue)\n\n./packstack/installer/run_setup.py:115:37: F821 undefined name \u0027raw_input\u0027\n                        userInput \u003d raw_input(message.read())\n                                    ^\n./packstack/installer/run_setup.py:194:15: F821 undefined name \u0027raw_input\u0027\n        raw \u003d raw_input(message.read())\n              ^\n\nChange-Id: Ie100fa47472c2bd55635d5b4a524cf9ef6671940\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/aaef82ada428f299a6c529917bea905fce1edaa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/aaef82ada428f299a6c529917bea905fce1edaa8"}]},"branch":"refs/heads/master"},"bd593b8df76db090a7bafc1032175475c90a25a3":{"kind":"REWORK","_number":2,"created":"2024-09-28 12:43:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/930807/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/07/930807/2","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/07/930807/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/07/930807/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/07/930807/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/07/930807/2"}}},"commit":{"parents":[{"commit":"6253bb59ce168b841900049ab68531e89c1993eb","subject":"Merge \"Bump hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6253bb59ce168b841900049ab68531e89c1993eb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-28 12:11:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-28 12:16:30.000000000","tz":540},"subject":"Fix undefined names","message":"Fix undefined names\n\nThis fixes apparent problems detected when F821 check is enabled.\n\n./packstack/installer/core/drones.py:43:16: F821 undefined name \u0027ScriptRuntimeError\u0027\n        except ScriptRuntimeError as ex:\n               ^\n./packstack/installer/core/drones.py:313:9: F821 undefined name \u0027shutil\u0027\n        shutil.rmtree(self.local_tmpdir, ignore_errors\u003dTrue)\n\n./packstack/installer/run_setup.py:115:37: F821 undefined name \u0027raw_input\u0027\n                        userInput \u003d raw_input(message.read())\n                                    ^\n./packstack/installer/run_setup.py:194:15: F821 undefined name \u0027raw_input\u0027\n        raw \u003d raw_input(message.read())\n              ^\n./tests/installer/test_drones.py:82:23: F821 undefined name \u0027tarfile\u0027\n            tarball \u003d tarfile.open(pack_path)\n                      ^\n\nChange-Id: Ie100fa47472c2bd55635d5b4a524cf9ef6671940\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bd593b8df76db090a7bafc1032175475c90a25a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/bd593b8df76db090a7bafc1032175475c90a25a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
