)]}'
{"id":"x%2Fpackstack~931323","triplet_id":"x%2Fpackstack~master~I36731fea93867254eea96074050387b6689fe173","project":"x/packstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-04 11:04:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I36731fea93867254eea96074050387b6689fe173","subject":"tempest: Drop deprecated auth_version","status":"MERGED","created":"2024-10-03 16:48:59.000000000","updated":"2024-10-04 11:04:50.000000000","submitted":"2024-10-04 11:04:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931323","meta_rev_id":"5531aa2090d708e146a12172ee922a49d12d874a","_number":931323,"virtual_id_number":931323,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-04 11:04:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":2,"date":"2024-10-04 10:05:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"all":[{"value":1,"date":"2024-10-04 10:05:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-03 17:52:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-04 10:05:04.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"}],"messages":[{"id":"21b7bc21ebdb06beaf58688daadfe1668a616da9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-03 16:48:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"933f6567a501d0c094209e06166a5cf8f7a29d08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 17:52:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67651cac1305491799b580bff1140ede\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9f3813dce0949249c4061783be5c15e : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11bdf1d93f6c47f9998e84e966b53d3f : SUCCESS in 2m 43s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/c24cc236aa9a4f03a23267e82de9353a : SUCCESS in 51m 13s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/8d5b980e850c4c18859935d0d6945b7a : SUCCESS in 57m 51s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/b9b8d7562d1746f5847624328e6dd56d : SUCCESS in 31m 52s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/73b0952eb7814b068c7db3df9e8b41c3 : SUCCESS in 47m 10s","accounts_in_message":[],"_revision_number":1},{"id":"eff1cac2f80982451b323e063cf30b53dd0db14a","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-10-04 10:05:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2c7556bc54cf28479472ebd2ec3da093339cdcc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 10:05:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"73fcf92b35d47da7202d50ec726e4e99058e565f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 11:04:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c15e0be69f2c41e69259902f9e4476e0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70cce4534e0b40eba385d9533d71c821 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da2c297a24af442ebbf60dce9f2ae51b : SUCCESS in 3m 14s\n- packstack-centos9s-integration-scenario001 https://zuul.opendev.org/t/openstack/build/26d48abc49174d9187e0373df7f1c81a : SUCCESS in 53m 00s\n- packstack-centos9s-integration-scenario002 https://zuul.opendev.org/t/openstack/build/3af0db4b2aa8415bba6dc42b9b7043d0 : SUCCESS in 52m 53s\n- packstack-centos9s-integration-scenario003 https://zuul.opendev.org/t/openstack/build/7d8cfd218e214564b452c0cd23c7ed83 : SUCCESS in 40m 54s\n- packstack-centos9s-multinode-scenario002-tempest https://zuul.opendev.org/t/openstack/build/96a2ce5d155a4969a6beea226995bf27 : SUCCESS in 43m 36s","accounts_in_message":[],"_revision_number":1},{"id":"5531aa2090d708e146a12172ee922a49d12d874a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 11:04:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c8838b098948cdeb2b2710cc0326524378b8d3f","revisions":{"6c8838b098948cdeb2b2710cc0326524378b8d3f":{"kind":"REWORK","_number":1,"created":"2024-10-03 16:48:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/931323/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/x/packstack","ref":"refs/changes/23/931323/1","commands":{"Checkout":"git fetch https://review.opendev.org/x/packstack refs/changes/23/931323/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/x/packstack refs/changes/23/931323/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/x/packstack refs/changes/23/931323/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/x/packstack refs/changes/23/931323/1"}}},"commit":{"parents":[{"commit":"6253bb59ce168b841900049ab68531e89c1993eb","subject":"Merge \"Bump hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6253bb59ce168b841900049ab68531e89c1993eb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-03 16:46:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-03 16:48:48.000000000","tz":540},"subject":"tempest: Drop deprecated auth_version","message":"tempest: Drop deprecated auth_version\n\nIt was deprecated in tempest 40.0.0[1]. The corresponding parameter of\npuppet-tempest was also deprecated[2].\n\n[1] https://github.com/openstack/tempest/commit/c0c90279ce\n[2] https://github.com/openstack/puppet-tempest/commit/1fd2598f\n\nChange-Id: I36731fea93867254eea96074050387b6689fe173\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6c8838b098948cdeb2b2710cc0326524378b8d3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/x/packstack/commit/6c8838b098948cdeb2b2710cc0326524378b8d3f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
